)]}'
{"id":"openstack%2Fcharm-neutron-openvswitch~808930","triplet_id":"openstack%2Fcharm-neutron-openvswitch~stable%2F21.04~I25c4be529e49fe3fed2b045e2c6bc117acd324c1","project":"openstack/charm-neutron-openvswitch","branch":"stable/21.04","topic":"bug/1934129","hashtags":[],"change_id":"I25c4be529e49fe3fed2b045e2c6bc117acd324c1","subject":"Add extension fwaas_v2 based on neutron-api relation data","status":"ABANDONED","created":"2021-09-14 11:18:20.000000000","updated":"2021-09-14 11:57:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a392c9adc52de212136ef750d149d9338a478f89","_number":808930,"virtual_id_number":808930,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-14 11:25:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 11:25:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0f0688646aa7a3fc9337d5393e71dd9bf1afffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-14 11:18:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fc3f306085408933707bc16ef9e0aeb1fe4c2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 11:25:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc4b5d23ac84faca0edd291fa6280f7 : SUCCESS in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/25f9c7ed4b31475daee71982f2c7034c : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff5003fe771b40619a729bf6bded6248 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/214dbef5e5a44845b783e7c45fe88236 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8d6d9cf1cba4f8f929af8bb39bdd431 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a392c9adc52de212136ef750d149d9338a478f89","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-14 11:57:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e4d619bfd3444de69e0902d3a03a243387f8d79","revisions":{"1e4d619bfd3444de69e0902d3a03a243387f8d79":{"kind":"REWORK","_number":1,"created":"2021-09-14 11:18:20.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/30/808930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/30/808930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/30/808930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/30/808930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/30/808930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/30/808930/1"}}},"commit":{"parents":[{"commit":"b5870d73c7f059c5a94850c8a2f03caa06973344","subject":"Set NoopFirewallDriver when SGs disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/b5870d73c7f059c5a94850c8a2f03caa06973344"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-09-14 11:11:51.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-09-14 11:18:16.000000000","tz":330},"subject":"Add extension fwaas_v2 based on neutron-api relation data","message":"Add extension fwaas_v2 based on neutron-api relation data\n\nfwaas_v2 extension is added in l3_agent.ini by default from stein.\nHowever with the newly introduced neutron-api configuration parameter\nenable-fwaas, the extension have to added only when it is set to true\non neutron-api charm.\n\nneutron-api charm adds enabled-fwaas as relation data in the commit\nhttps://review.opendev.org/c/openstack/charm-neutron-api/+/806676.\nThis patch removes special handling of fwaas_v2 as l3_extension_plugins\nis already updated in relation data based on neutron-api enable-fwaas\nconfiguration.\n\nAdd python3-neutron-fwaas in py3 package list only for rocky release. The\npackage is already part of dependency of neutron-l3-agent until Ussuri\nrelease and the dependency will be removed from Victoria release in\npackage control files.\n\nSynced charmhelpers to get related changes\nhttps://github.com/juju/charm-helpers/pull/640\n\nCloses-Bug: #1934129\nChange-Id: I1546f45635bfd7af76001c1df4f99d219a9d8039\n(cherry picked from d3daa09bc49bcc5c66791565e1720a9a484ea149)\n\nChange-Id: I25c4be529e49fe3fed2b045e2c6bc117acd324c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/1e4d619bfd3444de69e0902d3a03a243387f8d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/1e4d619bfd3444de69e0902d3a03a243387f8d79"}]},"branch":"refs/heads/stable/21.04"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
