)]}'
{"id":"openstack%2Fcharm-neutron-openvswitch~808937","triplet_id":"openstack%2Fcharm-neutron-openvswitch~stable%2F21.04~I1546f45635bfd7af76001c1df4f99d219a9d8039","project":"openstack/charm-neutron-openvswitch","branch":"stable/21.04","topic":"bug/1934129","hashtags":[],"change_id":"I1546f45635bfd7af76001c1df4f99d219a9d8039","subject":"Add extension fwaas_v2 based on neutron-api relation data","status":"MERGED","created":"2021-09-14 12:22:20.000000000","updated":"2021-09-20 11:11:26.000000000","submitted":"2021-09-20 08:24:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808937-bug/1934129","meta_rev_id":"dd8049ec6564d9bbca2eb6d634a782b051b3f4b1","_number":808937,"virtual_id_number":808937,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":-1,"date":"2021-09-14 16:57:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 08:24:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":2,"date":"2021-09-20 08:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-20 08:00:25.000000000","_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":1,"date":"2021-09-20 08:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-09-20 06:48:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 12:34:06.000000000","updated_by":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"reviewer":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"state":"CC"},{"updated":"2021-09-14 12:37:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 12:40:19.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"},{"updated":"2021-09-14 16:57:25.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 08:00:25.000000000","updated_by":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"reviewer":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"state":"REVIEWER"},{"updated":"2021-09-20 08:16:37.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"}],"messages":[{"id":"0ad6153f92367b2090efe8f2a03de7caccc1dc10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-14 12:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a199ab151678a68df7e75b34cd7262998a0ac4f8","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-09-14 12:34:06.000000000","message":"Patch Set 1:\n\nThanks for the patch! Has there been a discussion already about whether or not we are fine taking this as a stable backport or would I be the first one to make the call?","accounts_in_message":[],"_revision_number":1},{"id":"f400301fec401ba11db7ef6c2fb07e2f40574b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 12:37:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2764dd5462bd492a99483597e1bae359 : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7fb0aa49398f46b7a5aa6232b8f633bf : SUCCESS in 10m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77f3962ae75a4ae18b6c9b18ac68c8f6 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ad393b17248418aa8ff5247200e98fb : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8986a29f6a0f4a65b94f16acccc3151c : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"de625d047d24f5cb9dacc257583ef8137d5be569","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-14 12:39:04.000000000","message":"Patch Set 1: Workflow-1\n\nDNM, this requires merge of https://github.com/juju/charm-helpers/pull/640","accounts_in_message":[],"_revision_number":1},{"id":"74ce152de3a0691b1b9481970bef239bc0ecf76f","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-14 16:57:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/346/808937/1/check/osci-lint/3462345/index.html : SUCCESS in 4m 00s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/97b/808937/1/check/tox-py35/97bdcd8/index.html : SUCCESS in 5m 55s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/d6b/808937/1/check/tox-py36/d6bd860/index.html : SUCCESS in 5m 55s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/f00/808937/1/check/tox-py37/f00a3e6/index.html : SUCCESS in 5m 40s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/83f/808937/1/check/tox-py38/83fb258/index.html : SUCCESS in 5m 56s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/2df/808937/1/check/xenial-mitaka_ovs/2dfbbc7/index.html : TIMED_OUT in 3h 05m 08s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/4ff/808937/1/check/bionic-queens-dvr-snat/4ff2f1c/index.html : SUCCESS in 1h 08m 12s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/87d/808937/1/check/bionic-stein-dvr-snat/87db427/index.html : SUCCESS in 1h 15m 16s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/782/808937/1/check/bionic-train-dvr-snat/7822529/index.html : SUCCESS in 1h 17m 29s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/6d2/808937/1/check/bionic-ussuri-dvr-snat/6d28e29/index.html : FAILURE in 39m 26s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/5be/808937/1/check/migrate-ovn-focal-ussuri-dvr-snat/5be5731/index.html : SUCCESS in 1h 11m 02s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/826/808937/1/check/migrate-ovn-focal-victoria-dvr-snat/8261433/index.html : SUCCESS in 1h 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4d088e773ea93852b23ffb3c216a45b753e80485","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-15 03:12:12.000000000","message":"Patch Set 1:\n\nzosci-recheck\n\nCI failures:\nxenial-mitaka_ovs       TIMED_OUT\nbionic-ussuri-dvr-snat  Not related to the patch, neutron-openvswitch waiting for restart","accounts_in_message":[],"_revision_number":1},{"id":"d61c1e4cdafd2a48d2d808c3ada688bb6cb24f70","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-15 07:31:23.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/362/808937/1/check/osci-lint/362d8f6/index.html : SUCCESS in 4m 29s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/5d6/808937/1/check/tox-py35/5d6fd3d/index.html : SUCCESS in 5m 46s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/b2b/808937/1/check/tox-py36/b2bec14/index.html : SUCCESS in 6m 10s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/fa1/808937/1/check/tox-py37/fa129d9/index.html : SUCCESS in 5m 59s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/393/808937/1/check/tox-py38/393446d/index.html : SUCCESS in 6m 15s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/fb7/808937/1/check/xenial-mitaka_ovs/fb7e076/index.html : TIMED_OUT in 3h 06m 19s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/f3f/808937/1/check/bionic-queens-dvr-snat/f3fa25a/index.html : SUCCESS in 1h 07m 34s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d0d/808937/1/check/bionic-stein-dvr-snat/d0d52e5/index.html : SUCCESS in 1h 14m 37s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/c2a/808937/1/check/bionic-train-dvr-snat/c2aeb62/index.html : SUCCESS in 1h 18m 19s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/512/808937/1/check/bionic-ussuri-dvr-snat/512d220/index.html : SUCCESS in 1h 16m 23s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/2b1/808937/1/check/migrate-ovn-focal-ussuri-dvr-snat/2b19d7b/index.html : SUCCESS in 1h 04m 27s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/894/808937/1/check/migrate-ovn-focal-victoria-dvr-snat/8943b9b/index.html : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2e6e05e20e87a7470b37899620e6f8f6919c83b9","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-15 08:28:20.000000000","message":"Patch Set 1:\n\nxenial-mitaka_ovs failure is unrelated, see bug https://bugs.launchpad.net/charm-neutron-openvswitch/+bug/1936379","accounts_in_message":[],"_revision_number":1},{"id":"0ecc1693ea472260c3422ac7c9574278559be3b4","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-15 09:59:56.000000000","message":"Patch Set 1: -Workflow\n\nReady for review","accounts_in_message":[],"_revision_number":1},{"id":"98b107f772df674b7fcbfeca4eb0be426d3a35e3","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-20 02:21:38.000000000","message":"Patch Set 1:\n\nzosci-recheck \n\nCI verification trigger after merge of related charm-neutron-api patch https://review.opendev.org/c/openstack/charm-neutron-api/+/808929","accounts_in_message":[],"_revision_number":1},{"id":"f7811b8f3e834df953b745fa29a851dc4a80a965","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-20 06:42:44.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/af9/808937/1/check/osci-lint/af9e223/index.html : SUCCESS in 4m 26s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/bd5/808937/1/check/tox-py35/bd5f609/index.html : SUCCESS in 6m 29s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/921/808937/1/check/tox-py36/9216d0d/index.html : SUCCESS in 6m 11s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/27f/808937/1/check/tox-py37/27f07c1/index.html : SUCCESS in 6m 02s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/4f3/808937/1/check/tox-py38/4f3d20c/index.html : SUCCESS in 6m 11s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/691/808937/1/check/xenial-mitaka_ovs/691d143/index.html : TIMED_OUT in 3h 05m 23s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/bbd/808937/1/check/bionic-queens-dvr-snat/bbdecf6/index.html : SUCCESS in 1h 10m 03s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/b86/808937/1/check/bionic-stein-dvr-snat/b86c4c4/index.html : SUCCESS in 1h 18m 52s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/574/808937/1/check/bionic-train-dvr-snat/5745590/index.html : SUCCESS in 1h 20m 36s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/7ff/808937/1/check/bionic-ussuri-dvr-snat/7ffbf33/index.html : FAILURE in 40m 41s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/e8d/808937/1/check/migrate-ovn-focal-ussuri-dvr-snat/e8d7c23/index.html : SUCCESS in 1h 07m 31s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/6ad/808937/1/check/migrate-ovn-focal-victoria-dvr-snat/6ad2e87/index.html : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"343aa17217412bda5c1b9aa2e54f459d2e137bfa","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-20 06:48:16.000000000","message":"Patch Set 1:\n\nzosci-recheck unrelated failure in CI job bionic-ussuri-dvr-snat","accounts_in_message":[],"_revision_number":1},{"id":"3fc4518d09f42096d46c1e26829fb556e6d438ca","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-09-20 08:00:25.000000000","message":"Patch Set 1: Code-Review+2\n\nAll bundles (but Mitaka as explained) passed in an earlier OSCI run so I think we\u0027re good in term of CI.","accounts_in_message":[],"_revision_number":1},{"id":"fc0f285d9af59c9a3121a209516d2aa8b41e871d","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2021-09-20 08:16:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce172717d056df7065e3f0ad1d39e3084c3accd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:17:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfb9fe3e185ff2c8791fe805c35a895c90ba52c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:24:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8e543dbdd548e4888c8967a2a79454 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa51cce2cfe046de9f045f097f1d552d : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c098458b2cb54cde9769808e52faa427 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43318c3d97944a4faf324e9e163747b1 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6ccd453d12c17269bb1f14ffd57ba4a12b4bc22c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:24:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dd8049ec6564d9bbca2eb6d634a782b051b3f4b1","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-20 11:11:26.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/cd5/808937/1/check/osci-lint/cd59fe3/index.html : SUCCESS in 3m 45s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/5ea/808937/1/check/tox-py35/5ea4763/index.html : SUCCESS in 5m 09s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/e9c/808937/1/check/tox-py36/e9ca73e/index.html : SUCCESS in 5m 39s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/dbb/808937/1/check/tox-py37/dbb3f86/index.html : SUCCESS in 5m 48s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/986/808937/1/check/tox-py38/986aeda/index.html : SUCCESS in 5m 48s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/48c/808937/1/check/xenial-mitaka_ovs/48c0bc6/index.html : TIMED_OUT in 3h 04m 56s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/531/808937/1/check/bionic-queens-dvr-snat/5319747/index.html : SUCCESS in 1h 02m 07s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/0af/808937/1/check/bionic-stein-dvr-snat/0afe9b8/index.html : SUCCESS in 1h 10m 26s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/7cb/808937/1/check/bionic-train-dvr-snat/7cb7ab5/index.html : SUCCESS in 1h 15m 08s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/2d6/808937/1/check/bionic-ussuri-dvr-snat/2d6174a/index.html : SUCCESS in 1h 31m 22s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/323/808937/1/check/migrate-ovn-focal-ussuri-dvr-snat/32311ab/index.html : SUCCESS in 1h 10m 50s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/2af/808937/1/check/migrate-ovn-focal-victoria-dvr-snat/2afbbe0/index.html : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efac9bd199eb0a03abfd247f363797a46ade45b1","revisions":{"efac9bd199eb0a03abfd247f363797a46ade45b1":{"kind":"REWORK","_number":1,"created":"2021-09-14 12:22:20.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/37/808937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/37/808937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/37/808937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/37/808937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/37/808937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/37/808937/1"}}},"commit":{"parents":[{"commit":"b5870d73c7f059c5a94850c8a2f03caa06973344","subject":"Set NoopFirewallDriver when SGs disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/b5870d73c7f059c5a94850c8a2f03caa06973344"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-31 05:44:43.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-09-14 12:22:16.000000000","tz":330},"subject":"Add extension fwaas_v2 based on neutron-api relation data","message":"Add extension fwaas_v2 based on neutron-api relation data\n\nfwaas_v2 extension is added in l3_agent.ini by default from stein.\nHowever with the newly introduced neutron-api configuration parameter\nenable-fwaas, the extension have to added only when it is set to true\non neutron-api charm.\n\nneutron-api charm adds enabled-fwaas as relation data in the commit\nhttps://review.opendev.org/c/openstack/charm-neutron-api/+/806676.\nThis patch removes special handling of fwaas_v2 as l3_extension_plugins\nis already updated in relation data based on neutron-api enable-fwaas\nconfiguration.\n\nAdd python3-neutron-fwaas in py3 package list only for rocky release. The\npackage is already part of dependency of neutron-l3-agent until Ussuri\nrelease and the dependency will be removed from Victoria release in\npackage control files.\n\nSynced charmhelpers to get related changes\nhttps://github.com/juju/charm-helpers/pull/640\n\nCloses-Bug: #1934129\nChange-Id: I1546f45635bfd7af76001c1df4f99d219a9d8039\n(cherry picked from commit d3daa09bc49bcc5c66791565e1720a9a484ea149)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/efac9bd199eb0a03abfd247f363797a46ade45b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/efac9bd199eb0a03abfd247f363797a46ade45b1"}]},"branch":"refs/heads/stable/21.04"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
