)]}'
{"id":"openstack%2Fcharm-neutron-openvswitch~813407","triplet_id":"openstack%2Fcharm-neutron-openvswitch~master~I2ee90140f646170552fd3a638af2231ac9a38cad","project":"openstack/charm-neutron-openvswitch","branch":"master","topic":"bug/1931696","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2021-10-25 11:07:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ee90140f646170552fd3a638af2231ac9a38cad","subject":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"","status":"MERGED","created":"2021-10-11 10:23:46.000000000","updated":"2021-10-25 11:07:03.000000000","submitted":"2021-10-25 11:07:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813407-bug/1931696","meta_rev_id":"6e825a33c12c3827a3c7a815eafc4269e645d0c3","_number":813407,"virtual_id_number":813407,"owner":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":1,"date":"2021-10-12 08:41:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 11:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"recommended":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"all":[{"value":1,"date":"2021-10-13 08:08:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-25 10:49:43.000000000","_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-11 11:24:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"all":[{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-25 10:49:43.000000000","_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 10:46:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 11:24:04.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2021-10-11 16:13:15.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 08:08:13.000000000","updated_by":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"reviewer":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"state":"REVIEWER"},{"updated":"2021-10-25 10:49:43.000000000","updated_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"reviewer":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"state":"REVIEWER"}],"messages":[{"id":"537c399cce92593e03ffa5683dad552e41a793d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-10-11 10:23:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9530476d6994906f3a8e1865288c8a93d81960d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-10-11 10:25:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7d07d4436a14aecfbd1505e38ba1cfed2de6def6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 10:46:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7f024fa8d6349f59f3c40103a73f1b4 : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b40fe4d432704d8086c0ed363fd513cd : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b6ca879492d46518aba4807f738e2cf : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/689808a9f1024b9ea5976d6676413f7f : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f325d9d3cad4028b3df36d116e811fd : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/548f03a82c82492a812115082b31edf7 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e86cbce7b57e0f9710e7946e2003851ff99eed92","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-10-11 11:24:47.000000000","message":"Patch Set 2: Code-Review+1\n\nNeed to be backported to stable/21.10","accounts_in_message":[],"_revision_number":2},{"id":"fcb13718a0137e65b2939adbe4396116b15c4fae","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-10-11 16:13:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/296/813407/2/check/charm-build/296aef0/ : SUCCESS in 4m 39s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/582/813407/2/check/osci-lint/5823105/ : SUCCESS in 5m 00s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/c15/813407/2/check/tox-py35/c152c67/ : SUCCESS in 5m 46s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/478/813407/2/check/tox-py36/478f585/ : SUCCESS in 5m 24s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/b2e/813407/2/check/tox-py37/b2ed514/ : SUCCESS in 6m 46s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/b58/813407/2/check/tox-py38/b58b1e2/ : SUCCESS in 5m 03s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/164/813407/2/check/xenial-mitaka_ovs/1644040/ : SUCCESS in 1h 11m 14s (non-voting)\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/198/813407/2/check/bionic-queens-dvr-snat/19867d5/ : SUCCESS in 1h 40m 39s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/b9e/813407/2/check/bionic-stein-dvr-snat/b9e9056/ : SUCCESS in 1h 49m 48s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/392/813407/2/check/bionic-train-dvr-snat/392f49c/ : SUCCESS in 1h 32m 27s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/90f/813407/2/check/bionic-ussuri-dvr-snat/90ffdef/ : SUCCESS in 1h 27m 59s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/21e/813407/2/check/migrate-ovn-focal-ussuri-dvr-snat/21e6fb9/ : SUCCESS in 1h 25m 09s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d44/813407/2/check/migrate-ovn-focal-victoria-dvr-snat/d44ba2e/ : FAILURE in 35m 07s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/cf9/813407/2/check/migrate-ovn-focal-wallaby-dvr-snat/cf9e3d1/ : TIMED_OUT in 3h 15m 55s (non-voting)\n- migrate-ovn-focal-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/1f4/813407/2/check/migrate-ovn-focal-xena-dvr-snat/1f422ec/ : FAILURE in 1h 31m 24s (non-voting)\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/62a/813407/2/check/migrate-ovn-groovy-victoria-dvr-snat/62ad289/ : SUCCESS in 1h 29m 54s (non-voting)\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/f00/813407/2/check/migrate-ovn-hirsute-wallaby-dvr-snat/f00dc0b/ : TIMED_OUT in 3h 16m 26s (non-voting)\n- migrate-ovn-impish-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/dab/813407/2/check/migrate-ovn-impish-xena-dvr-snat/daba4a1/ : FAILURE in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7c9b1869ba1e77e3016c4888be793cd6fbc967f","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-10-12 03:10:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f6d1c222be652b497717449f4f3a91b02d661f8","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-10-12 08:41:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/2f7/813407/2/check/charm-build/2f7f29a/ : SUCCESS in 4m 18s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/6eb/813407/2/check/osci-lint/6eb69b5/ : SUCCESS in 4m 21s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/7d2/813407/2/check/tox-py35/7d25cb3/ : SUCCESS in 4m 55s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/e37/813407/2/check/tox-py36/e376b8d/ : SUCCESS in 5m 51s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/621/813407/2/check/tox-py37/62197d3/ : SUCCESS in 5m 10s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/b3b/813407/2/check/tox-py38/b3b3724/ : SUCCESS in 6m 11s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/633/813407/2/check/xenial-mitaka_ovs/633a6c2/ : TIMED_OUT in 3h 05m 52s (non-voting)\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/021/813407/2/check/bionic-queens-dvr-snat/0218123/ : SUCCESS in 1h 28m 22s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/661/813407/2/check/bionic-stein-dvr-snat/6611772/ : SUCCESS in 1h 40m 17s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/f06/813407/2/check/bionic-train-dvr-snat/f06a220/ : SUCCESS in 1h 39m 04s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/333/813407/2/check/bionic-ussuri-dvr-snat/3331f70/ : SUCCESS in 1h 32m 48s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/83e/813407/2/check/migrate-ovn-focal-ussuri-dvr-snat/83e60bf/ : SUCCESS in 1h 16m 05s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/97b/813407/2/check/migrate-ovn-focal-victoria-dvr-snat/97bbae2/ : SUCCESS in 1h 26m 32s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d36/813407/2/check/migrate-ovn-focal-wallaby-dvr-snat/d36b4e1/ : TIMED_OUT in 3h 15m 50s (non-voting)\n- migrate-ovn-focal-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/40c/813407/2/check/migrate-ovn-focal-xena-dvr-snat/40caa44/ : FAILURE in 1h 24m 34s (non-voting)\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/6c2/813407/2/check/migrate-ovn-groovy-victoria-dvr-snat/6c2aa10/ : SUCCESS in 1h 27m 04s (non-voting)\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/25f/813407/2/check/migrate-ovn-hirsute-wallaby-dvr-snat/25f35a2/ : TIMED_OUT in 3h 16m 08s (non-voting)\n- migrate-ovn-impish-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/8e6/813407/2/check/migrate-ovn-impish-xena-dvr-snat/8e60c19/ : FAILURE in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15cbe276e91a9c8b088e947250a7dec1e96cbadd","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2021-10-13 08:08:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"62165b80a72c349055ed466383afe53b416eea3d","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2021-10-25 10:49:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aaf4e6c2eb1922d122853bd4d06236e8f9e42e4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 10:57:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4cdb3d1d4cc2791c6d8e7735d2b379d788ce7eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:07:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb20ad0b14ee4041832db5776e2e80c2 : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/103df9c56e1b486aab658bd02f996f0c : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbf54ab9c66c43f2890649d4181841bf : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/859f8a0a48d545c09520b6044162282d : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6e825a33c12c3827a3c7a815eafc4269e645d0c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:07:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6fb737c1c88c4773bedf65810429224c031ab881","revisions":{"8f5c43bf4781601d49b5b9bad79bc19921c02fae":{"kind":"REWORK","_number":1,"created":"2021-10-11 10:23:46.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/07/813407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/07/813407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/1"}}},"commit":{"parents":[{"commit":"2230c1e60f328e369e69afb6ece6a57b658a200c","subject":"Add xena bundles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/2230c1e60f328e369e69afb6ece6a57b658a200c"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-10-11 10:23:29.000000000","tz":60},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-10-11 10:23:29.000000000","tz":60},"subject":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"","message":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"\n\nThis reverts commit 0f7d135507def07196aa98cc4cf9fd518641a369.\n\nChange-Id: I2ee90140f646170552fd3a638af2231ac9a38cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/8f5c43bf4781601d49b5b9bad79bc19921c02fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/8f5c43bf4781601d49b5b9bad79bc19921c02fae"}]},"branch":"refs/heads/master"},"6fb737c1c88c4773bedf65810429224c031ab881":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-11 10:25:03.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/07/813407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/07/813407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/07/813407/2"}}},"commit":{"parents":[{"commit":"2230c1e60f328e369e69afb6ece6a57b658a200c","subject":"Add xena bundles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/2230c1e60f328e369e69afb6ece6a57b658a200c"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-10-11 10:23:29.000000000","tz":60},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-10-11 10:24:40.000000000","tz":60},"subject":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"","message":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"\n\nThis reverts commit 0f7d135507def07196aa98cc4cf9fd518641a369.\n\nRelated-Bug: #1931696\nChange-Id: I2ee90140f646170552fd3a638af2231ac9a38cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/6fb737c1c88c4773bedf65810429224c031ab881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/6fb737c1c88c4773bedf65810429224c031ab881"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
