)]}'
{"id":"openstack%2Fcharm-neutron-openvswitch~813609","triplet_id":"openstack%2Fcharm-neutron-openvswitch~stable%2F21.10~I2ee90140f646170552fd3a638af2231ac9a38cad","project":"openstack/charm-neutron-openvswitch","branch":"stable/21.10","topic":"bug/1931696","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2021-10-25 11:05:34.000000000","reason":"Change was submitted"},"10366":{"account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"last_update":"2021-10-25 11:05:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ee90140f646170552fd3a638af2231ac9a38cad","subject":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"","status":"MERGED","created":"2021-10-12 11:07:09.000000000","updated":"2021-10-25 11:05:34.000000000","submitted":"2021-10-25 11:05:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813609-bug/1931696","meta_rev_id":"97d78c8aa18ded1c84ef65cb144f43c1f52a9abe","_number":813609,"virtual_id_number":813609,"owner":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2021-10-22 18:48:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 11:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":2,"date":"2021-10-25 10:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-25 10:49:30.000000000","_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-25 11:05:34.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-25 10:49:30.000000000","_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 11:27:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 11:48:47.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2021-10-12 16:32:08.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 10:36:47.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2021-10-22 13:01:27.000000000","updated_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"reviewer":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"state":"CC"},{"updated":"2021-10-25 10:49:30.000000000","updated_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"reviewer":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"state":"REVIEWER"}],"messages":[{"id":"a6ef9394043e31d9691c68bd1d0f82b7377ced3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-10-12 11:07:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c73bd7e5b71cdd3542a954085e98eecc8b32191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 11:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6b079f651d4dc0b2f15bbc4bbe5d1b : SUCCESS in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55a78dadb8d84dc5ab9ff4abe177814a : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d72b515fdb049cba74659ff9cba473d : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a5e490eaa7d4065bfec59725f1a3499 : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/203641bd313f4f83897b2f56b168b772 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4f77740ea1ca9cca29ad79be9ee7e8d67a7c642d","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-10-12 11:48:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"246ee722881f17116cf89769c02603e17e5e6451","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-10-12 16:32:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/6e2/813609/1/check/charm-build/6e27d41/ : SUCCESS in 4m 00s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/f32/813609/1/check/osci-lint/f32c6fa/ : SUCCESS in 4m 37s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/6f8/813609/1/check/tox-py35/6f88a05/ : SUCCESS in 4m 51s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/733/813609/1/check/tox-py36/7336a83/ : SUCCESS in 6m 02s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/41e/813609/1/check/tox-py37/41ebbf2/ : SUCCESS in 6m 09s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/3ed/813609/1/check/tox-py38/3edeb50/ : SUCCESS in 5m 22s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/67b/813609/1/check/xenial-mitaka_ovs/67b9866/ : TIMED_OUT in 3h 06m 42s (non-voting)\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/cba/813609/1/check/bionic-queens-dvr-snat/cba03b6/ : SUCCESS in 1h 19m 55s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/b79/813609/1/check/bionic-stein-dvr-snat/b79ae32/ : SUCCESS in 1h 39m 00s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/9ef/813609/1/check/bionic-train-dvr-snat/9ef6d4c/ : SUCCESS in 1h 43m 39s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/9c4/813609/1/check/bionic-ussuri-dvr-snat/9c4cdf7/ : FAILURE in 48m 08s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/6fd/813609/1/check/migrate-ovn-focal-ussuri-dvr-snat/6fd3958/ : SUCCESS in 1h 13m 32s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/70f/813609/1/check/migrate-ovn-focal-victoria-dvr-snat/70f146e/ : SUCCESS in 1h 28m 12s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/aa8/813609/1/check/migrate-ovn-focal-wallaby-dvr-snat/aa80b2f/ : TIMED_OUT in 3h 17m 39s (non-voting)\n- migrate-ovn-focal-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/90f/813609/1/check/migrate-ovn-focal-xena-dvr-snat/90f390e/ : TIMED_OUT in 3h 17m 08s (non-voting)\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/36a/813609/1/check/migrate-ovn-groovy-victoria-dvr-snat/36af712/ : SUCCESS in 1h 24m 40s (non-voting)\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/a1a/813609/1/check/migrate-ovn-hirsute-wallaby-dvr-snat/a1a10ae/ : TIMED_OUT in 3h 17m 04s (non-voting)\n- migrate-ovn-impish-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ec7/813609/1/check/migrate-ovn-impish-xena-dvr-snat/ec79af8/ : FAILURE in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fceffdf540714af8ad3697758e516b93398f8d34","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-10-13 02:55:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f38949bfb284f36d5e2253c1ec723d674738d02","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-10-13 08:53:49.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/dc9/813609/1/check/charm-build/dc959ad/ : SUCCESS in 5m 22s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/6b0/813609/1/check/osci-lint/6b0b8c8/ : SUCCESS in 5m 53s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/f75/813609/1/check/tox-py35/f75d038/ : SUCCESS in 6m 55s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/f23/813609/1/check/tox-py36/f2319a4/ : SUCCESS in 7m 31s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/077/813609/1/check/tox-py37/077869d/ : SUCCESS in 7m 35s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/a1a/813609/1/check/tox-py38/a1a54a1/ : SUCCESS in 7m 04s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/1af/813609/1/check/xenial-mitaka_ovs/1af54f4/ : SUCCESS in 1h 34m 13s (non-voting)\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ba1/813609/1/check/bionic-queens-dvr-snat/ba15810/ : SUCCESS in 1h 48m 20s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/328/813609/1/check/bionic-stein-dvr-snat/328ebab/ : SUCCESS in 1h 46m 27s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/288/813609/1/check/bionic-train-dvr-snat/288e089/ : SUCCESS in 2h 15m 22s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/74b/813609/1/check/bionic-ussuri-dvr-snat/74b5757/ : SUCCESS in 1h 49m 14s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/12c/813609/1/check/migrate-ovn-focal-ussuri-dvr-snat/12c0345/ : SUCCESS in 1h 27m 10s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/61f/813609/1/check/migrate-ovn-focal-victoria-dvr-snat/61fcc0a/ : FAILURE in 1h 03m 41s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/20d/813609/1/check/migrate-ovn-focal-wallaby-dvr-snat/20d598e/ : FAILURE in 1h 02m 33s (non-voting)\n- migrate-ovn-focal-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/b5d/813609/1/check/migrate-ovn-focal-xena-dvr-snat/b5d02d0/ : TIMED_OUT in 3h 21m 15s (non-voting)\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/32e/813609/1/check/migrate-ovn-groovy-victoria-dvr-snat/32e8d60/ : FAILURE in 59m 28s (non-voting)\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/82e/813609/1/check/migrate-ovn-hirsute-wallaby-dvr-snat/82ebb26/ : TIMED_OUT in 3h 22m 55s (non-voting)\n- migrate-ovn-impish-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/5c9/813609/1/check/migrate-ovn-impish-xena-dvr-snat/5c929e2/ : FAILURE in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0460d6ebd4b4b25572daa5eb008610375cdde123","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-10-13 10:36:47.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ec0d43049b025dedb680b052827b83ed7ebacb6","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-10-18 13:21:02.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"388a84827390651b32bb7689a5e36a8bfd650856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 13:41:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b1210bb9ce413787acdded90ae5648 : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8499b14ce6974bb7ab522ddcb9e9b2bb : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/952dd10b358346b081d13787eef157e1 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85310621baf74e7f8a4bc82d123db34d : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1e535ee0e354f5999ca22ad18c15df1 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5ed67d8a0d127a6e5ca0687aeaad17d99193967a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-10-18 19:35:44.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/d17/813609/1/check/charm-build/d17e586/ : SUCCESS in 4m 24s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/c18/813609/1/check/osci-lint/c18e647/ : SUCCESS in 3m 59s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/d23/813609/1/check/tox-py35/d2347d5/ : SUCCESS in 6m 09s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/ee8/813609/1/check/tox-py36/ee89606/ : SUCCESS in 6m 30s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/388/813609/1/check/tox-py37/388c099/ : SUCCESS in 6m 58s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/691/813609/1/check/tox-py38/691777e/ : SUCCESS in 5m 35s\n- xenial-mitaka_ovs finger://juju-22c89a-zuul-12.project.serverstack/afe477c29e57420aa3442f83e9780b67 : RETRY_LIMIT in 1s (non-voting)\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/89f/813609/1/check/bionic-queens-dvr-snat/89f688a/ : SUCCESS in 1h 30m 45s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/c24/813609/1/check/bionic-stein-dvr-snat/c24b6b4/ : SUCCESS in 2h 27m 35s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/3fb/813609/1/check/bionic-train-dvr-snat/3fb6acd/ : SUCCESS in 2h 03m 52s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/714/813609/1/check/bionic-ussuri-dvr-snat/7148ba7/ : FAILURE in 17m 11s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ceb/813609/1/check/migrate-ovn-focal-ussuri-dvr-snat/ceb905d/ : SUCCESS in 1h 34m 21s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/5a6/813609/1/check/migrate-ovn-focal-victoria-dvr-snat/5a6a658/ : SUCCESS in 2h 05m 18s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/586/813609/1/check/migrate-ovn-focal-wallaby-dvr-snat/5866afb/ : FAILURE in 1h 36m 25s (non-voting)\n- migrate-ovn-focal-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ef3/813609/1/check/migrate-ovn-focal-xena-dvr-snat/ef31544/ : TIMED_OUT in 3h 16m 09s (non-voting)\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/182/813609/1/check/migrate-ovn-groovy-victoria-dvr-snat/182618a/ : SUCCESS in 1h 50m 51s (non-voting)\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/04b/813609/1/check/migrate-ovn-hirsute-wallaby-dvr-snat/04b5aee/ : SUCCESS in 2h 18m 31s (non-voting)\n- migrate-ovn-impish-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/dbe/813609/1/check/migrate-ovn-impish-xena-dvr-snat/dbef6ab/ : FAILURE in 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"785f879d283d047a8026bc2b07584c58688112a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-10-21 07:56:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"85ed49c0d27496ed08d3003bf5923dadd1e6c4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 08:13:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb98f7397654a919abe1b62a7753ca6 : SUCCESS in 4m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b61dba05b3a4768b20952a8cdb22f1a : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/918790aa71554b188c9cafa48dfa62fb : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a60d008658541e88c24be77bbaf7776 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51ebefdc0a0c446ea40952a0f5fc8311 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":2},{"id":"60ce860469af7bb4201a25f5c4e1788926c934af","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2021-10-22 13:01:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4d5754a1fe8ef2f041ad4561eec1a2bb603c264","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-10-22 18:48:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/15c/813609/2/check/charm-build/15c68fe/ : SUCCESS in 3m 35s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/390/813609/2/check/osci-lint/3908dcd/ : SUCCESS in 4m 11s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/c51/813609/2/check/tox-py35/c51cc46/ : SUCCESS in 4m 35s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/e74/813609/2/check/tox-py36/e74d8b3/ : SUCCESS in 5m 43s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/a9a/813609/2/check/tox-py37/a9a6847/ : SUCCESS in 5m 21s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/6d0/813609/2/check/tox-py38/6d077c1/ : SUCCESS in 5m 13s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/144/813609/2/check/xenial-mitaka_ovs/1445a09/ : TIMED_OUT in 3h 05m 41s (non-voting)\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ae7/813609/2/check/bionic-queens-dvr-snat/ae714c3/ : SUCCESS in 1h 19m 21s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/4e2/813609/2/check/bionic-stein-dvr-snat/4e228aa/ : SUCCESS in 1h 29m 13s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ad0/813609/2/check/bionic-train-dvr-snat/ad07726/ : SUCCESS in 1h 17m 41s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d2a/813609/2/check/bionic-ussuri-dvr-snat/d2a8b04/ : SUCCESS in 1h 16m 11s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/714/813609/2/check/migrate-ovn-focal-ussuri-dvr-snat/714b18a/ : SUCCESS in 1h 11m 48s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/787/813609/2/check/migrate-ovn-focal-victoria-dvr-snat/787357e/ : SUCCESS in 1h 15m 10s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/0f8/813609/2/check/migrate-ovn-focal-wallaby-dvr-snat/0f8f3ee/ : TIMED_OUT in 3h 15m 53s (non-voting)\n- migrate-ovn-focal-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/cbe/813609/2/check/migrate-ovn-focal-xena-dvr-snat/cbef5b9/ : FAILURE in 26m 25s (non-voting)\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/30b/813609/2/check/migrate-ovn-groovy-victoria-dvr-snat/30b7431/ : SUCCESS in 1h 10m 12s (non-voting)\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/b3f/813609/2/check/migrate-ovn-hirsute-wallaby-dvr-snat/b3f0c07/ : TIMED_OUT in 3h 14m 53s (non-voting)\n- migrate-ovn-impish-xena-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/4cc/813609/2/check/migrate-ovn-impish-xena-dvr-snat/4cc5f40/ : FAILURE in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fa042a3ad05c76f0f999164afbf31623e54b9f1","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-10-25 10:48:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f83c2ec82fc77c183d0b0071fd611f9fc2faad12","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2021-10-25 10:49:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2aa92eec8f6c1a795869a2fe68f2c0c3a592e82e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 10:57:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"384b2ab6f3e30a30e98bfbe06ab071eca99a056f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:05:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc7ac8749c445879ae23c4d9a24298c : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4abed50bd2ee48b2b7d3a1fab4fb15b9 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e61f30aa99cf458b887bfd078a976191 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f94f524345945d991b9209c5376dd9b : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":2},{"id":"9cc456657d32877814e70902bdc886019d0d4f64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:05:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d465376858a7a88c1ccf4ea41f3c17fe5ceffd91","revisions":{"1e259c3fca9c54852867a2813d9e26648b568aa8":{"kind":"REWORK","_number":1,"created":"2021-10-12 11:07:09.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/09/813609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/09/813609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/1"}}},"commit":{"parents":[{"commit":"2230c1e60f328e369e69afb6ece6a57b658a200c","subject":"Add xena bundles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/2230c1e60f328e369e69afb6ece6a57b658a200c"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-10-11 10:23:29.000000000","tz":60},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-10-12 11:06:58.000000000","tz":60},"subject":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"","message":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"\n\nThis reverts commit 0f7d135507def07196aa98cc4cf9fd518641a369.\n\nRelated-Bug: #1931696\nChange-Id: I2ee90140f646170552fd3a638af2231ac9a38cad\n(cherry picked from commit 6fb737c1c88c4773bedf65810429224c031ab881)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/1e259c3fca9c54852867a2813d9e26648b568aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/1e259c3fca9c54852867a2813d9e26648b568aa8"}]},"branch":"refs/heads/stable/21.10"},"d465376858a7a88c1ccf4ea41f3c17fe5ceffd91":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-21 07:56:31.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/09/813609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/09/813609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/09/813609/2"}}},"commit":{"parents":[{"commit":"5317860eeb59789fa55b2b2f86850fd46916d1ef","subject":"21.10 - Stable cut of charms for testing period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/5317860eeb59789fa55b2b2f86850fd46916d1ef"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-10-11 10:23:29.000000000","tz":60},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2021-10-21 07:56:31.000000000","tz":0},"subject":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"","message":"Revert \"Set explicitly_egress_direct\u003dTrue for ml2 ovs\"\n\nThis reverts commit 0f7d135507def07196aa98cc4cf9fd518641a369.\n\nRelated-Bug: #1931696\nChange-Id: I2ee90140f646170552fd3a638af2231ac9a38cad\n(cherry picked from commit 6fb737c1c88c4773bedf65810429224c031ab881)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/d465376858a7a88c1ccf4ea41f3c17fe5ceffd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/d465376858a7a88c1ccf4ea41f3c17fe5ceffd91"}]},"branch":"refs/heads/stable/21.10","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
