)]}'
{"id":"openstack%2Fcharm-nova-compute~639719","triplet_id":"openstack%2Fcharm-nova-compute~master~I4fabe34112f1dedc0362edcded249ea82455d941","project":"openstack/charm-nova-compute","branch":"master","topic":"bug/1815844","hashtags":[],"change_id":"I4fabe34112f1dedc0362edcded249ea82455d941","subject":"config: add note for deployers when multipath is configured","status":"ABANDONED","created":"2019-02-27 15:56:52.000000000","updated":"2023-06-16 14:02:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"93ea0ac0777788058f10ec783c3889d64e58a975","_number":639719,"virtual_id_number":639719,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2019-03-05 08:44:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2019-02-27 16:12:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"all":[{"value":-1,"date":"2019-03-06 13:50:34.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-27 16:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-05 08:44:27.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"899091e8fd93fb59e602bb621f528ed8edd5da3d","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2019-02-27 15:56:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de301f09aa341098d6d50f2801d341ef3e00c007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 16:12:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/19/639719/1/check/openstack-tox-pep8/24556bd/ : SUCCESS in 6m 16s\n- openstack-tox-py35 http://logs.openstack.org/19/639719/1/check/openstack-tox-py35/b110351/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"60bc69d3b7a5a01771e509940d29341e68c332cb","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-02-27 16:13:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/8793/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/8793/index.html : FAILURE\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/8793/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/8793/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"d1b7b91ad34289745cb057cc2887bcafaf15b821","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2019-03-05 06:50:01.000000000","message":"Patch Set 1:\n\ncharm-recheck","accounts_in_message":[],"_revision_number":1},{"id":"110a961cb77dcfbe4b81d16e93c9a8a5e42bec92","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-03-05 08:44:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/9039/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/9039/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/9039/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/639719/1/9039/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"3b0514039d05ddec86aea6241766ef250f3ab6fb","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2019-03-06 13:50:34.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s see first whether os-brick community wants to fix the issue:\n\n  https://review.openstack.org/#/c/638639/","accounts_in_message":[],"_revision_number":1},{"id":"93ea0ac0777788058f10ec783c3889d64e58a975","tag":"autogenerated:gerrit:abandon","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-06-16 14:02:35.000000000","message":"Abandoned\n\nAbandoning as submitter put a hold on it but hasn\u0027t revisited in 4 years.  Please feel free to re-open it if it is still valid.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b428c913255dcd7f69372c36d3bba153af795b87","revisions":{"b428c913255dcd7f69372c36d3bba153af795b87":{"kind":"REWORK","_number":1,"created":"2019-02-27 15:56:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/19/639719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/19/639719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/19/639719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/19/639719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/19/639719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/19/639719/1"}}},"commit":{"parents":[{"commit":"9254beb043bb73e914430125eceec03afd8ea855","subject":"Update rabbit driver config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/9254beb043bb73e914430125eceec03afd8ea855"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@canonical.com","date":"2019-02-27 15:52:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@canonical.com","date":"2019-02-27 15:52:10.000000000","tz":60},"subject":"config: add note for deployers when multipath is configured","message":"config: add note for deployers when multipath is configured\n\nThe current release of os-brick does not support find multipaths\nfeature. This commit adds small note about it.\n\nChange-Id: I4fabe34112f1dedc0362edcded249ea82455d941\nCloses-Bug: 1815844\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/b428c913255dcd7f69372c36d3bba153af795b87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/b428c913255dcd7f69372c36d3bba153af795b87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
