)]}'
{"id":"openstack%2Fcharm-nova-compute~658062","triplet_id":"openstack%2Fcharm-nova-compute~master~I9d5e7c2ad034cd7bdf9cbf292b9107577860dac1","project":"openstack/charm-nova-compute","branch":"master","topic":"bug/1828259","hashtags":[],"change_id":"I9d5e7c2ad034cd7bdf9cbf292b9107577860dac1","subject":"py3: deal with more subordinate dependencies","status":"MERGED","created":"2019-05-09 12:19:19.000000000","updated":"2019-05-13 08:43:20.000000000","submitted":"2019-05-10 15:36:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658062-1557502562508-0f9a02b3","meta_rev_id":"d92b1bf4a3bdf4fed297eb5a6c53787e92055ff9","_number":658062,"virtual_id_number":658062,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-09 13:03:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":2,"date":"2019-05-10 15:36:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-10 15:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-10 15:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 13:03:15.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-10 15:25:02.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"state":"REVIEWER"},{"updated":"2019-05-10 15:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4dff0cf1b02915e6fa993c5700e59331f5e3561","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-05-09 12:19:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"207d5525ef09ab2e025a3792e6fac8d2cdc39d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:30:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/658062/1/check/openstack-tox-pep8/06ac150/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/62/658062/1/check/openstack-tox-py35/f455657/ : SUCCESS in 4m 41s\n- openstack-tox-cover http://logs.openstack.org/62/658062/1/check/openstack-tox-cover/cf7eef3/cover/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fa3cd3192882ace0af9803fb873c82bacc0e5013","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 13:03:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/658062/1/10277/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/658062/1/10277/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/658062/1/10277/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-nova-compute/658062/1/10277/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"46d600589e02686274ed81de80fad80a790df040","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2019-05-10 15:25:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be0182788605c90c80eed9f167f42d752aca21a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:25:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eeaaba818876f586a71d8b213d1cfa94499317a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:36:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"532ab638ee641e8ff4b0d677acd5003ebb5fdce5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:36:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/658062/1/gate/openstack-tox-pep8/16023b5/ : SUCCESS in 7m 55s\n- openstack-tox-py35 http://logs.openstack.org/62/658062/1/gate/openstack-tox-py35/e546300/ : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"389a3b71a713e0c6e8887b54241542b3b52714cd","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2019-05-13 07:30:08.000000000","message":"Patch Set 1:\n\n@Dmitrii - stable backports typically require 2 reviewers to approve them","accounts_in_message":[],"_revision_number":1},{"id":"809011699c644ddf80a8fc9df9ec40ad1bd9608d","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2019-05-13 08:43:20.000000000","message":"Patch Set 1:\n\nChris, agreed. But this one is for the master branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96b5e8a2493305ec7667c153d070dffda1bf547b","revisions":{"96b5e8a2493305ec7667c153d070dffda1bf547b":{"kind":"REWORK","_number":1,"created":"2019-05-09 12:19:19.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/62/658062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/62/658062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/62/658062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/62/658062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/62/658062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/62/658062/1"}}},"commit":{"parents":[{"commit":"4e9bcbc28b9e7dca66c9b22d886ac5c1979eb73e","subject":"Merge \"config: add notification-format option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/4e9bcbc28b9e7dca66c9b22d886ac5c1979eb73e"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-05-09 12:13:51.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-05-09 12:19:18.000000000","tz":-240},"subject":"py3: deal with more subordinate dependencies","message":"py3: deal with more subordinate dependencies\n\nEnsure subordinate py3 packages are installed if their py2\ncounter-parts are currently installed for neutron-openvswitch.\nIn this case we had missed python(3)-neutron-fwaas.\n\nChange-Id: I9d5e7c2ad034cd7bdf9cbf292b9107577860dac1\nCloses-Bug: #1828259\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/96b5e8a2493305ec7667c153d070dffda1bf547b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/96b5e8a2493305ec7667c153d070dffda1bf547b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
