)]}'
{"id":"openstack%2Fcharm-nova-compute~746089","triplet_id":"openstack%2Fcharm-nova-compute~stable%2F20.08~I2f91405f170577c43e1f6a503c10add93b694909","project":"openstack/charm-nova-compute","branch":"stable/20.08","topic":"release-20.08","hashtags":[],"change_id":"I2f91405f170577c43e1f6a503c10add93b694909","subject":"Updates for stable branch creation","status":"MERGED","created":"2020-08-13 10:37:25.000000000","updated":"2020-08-13 12:15:17.000000000","submitted":"2020-08-13 12:15:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746089-1597320917247-5080a3ed","meta_rev_id":"91b02e7ebe83257149dc732d3dc0d974b1b727c6","_number":746089,"virtual_id_number":746089,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 12:15:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"all":[{"value":2,"date":"2020-08-13 12:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"all":[{"value":1,"date":"2020-08-13 12:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 12:07:53.000000000","updated_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"reviewer":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"state":"REVIEWER"},{"updated":"2020-08-13 12:15:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"662347377e25d7be47c6424089378bae4e10fcd6","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-08-13 10:37:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff33a19ca9cb8c67083dab562a78270fb39c7a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 10:48:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ddade8c1f064d3595666e3aa349d1d3 : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e1cd6ef98cdd4c8194308f54883ddfc9 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/209c33b528b04db8a220bc65e9e5f7b1 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2f5e8bdb3af4146b88d3da552b39d26 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d891448b3abb40f18a9194f6063fe075 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2445bf2df001ddc1f37e430906a79b5c1f8722de","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2020-08-13 12:07:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"8620b358a07fa9ca63e03472216a49abc7cc42e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:08:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dbc9e6470b1f5e43315f0c74b5012c72b408fda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:15:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"471ffee084901768137b3732c2ed7705df410ec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:15:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0aae3bce0e4415ab33647a8650492d : SUCCESS in 3m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/19a2eaf2747745b1a09deb67a4b3109a : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31ba217ba63f45e09406c2afaf3e5814 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03455536dcef42faa2d70e0a6a6b9967 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67bd776d9e71ffa30841cce29ea1d5e25179fee8","revisions":{"67bd776d9e71ffa30841cce29ea1d5e25179fee8":{"kind":"REWORK","_number":1,"created":"2020-08-13 10:37:25.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/89/746089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/89/746089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/89/746089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/89/746089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/89/746089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/89/746089/1"}}},"commit":{"parents":[{"commit":"7dea8da5fbb8b78099937c00ceac868f8428ab8d","subject":"Release sync for 20.08","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/7dea8da5fbb8b78099937c00ceac868f8428ab8d"}]}],"author":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:37:13.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:37:23.000000000","tz":60},"subject":"Updates for stable branch creation","message":"Updates for stable branch creation\n\nSet default branch for git review/gerrit.\n\nSwitch tests to stable.\n\nSwitch to using stable charm-helpers branch.\n\nSwitch to using stable charm.openstack branch.\n\nChange-Id: I2f91405f170577c43e1f6a503c10add93b694909\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/67bd776d9e71ffa30841cce29ea1d5e25179fee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/67bd776d9e71ffa30841cce29ea1d5e25179fee8"}]},"branch":"refs/heads/stable/20.08"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
