)]}'
{"id":"openstack%2Fcharm-nova-compute~847582","triplet_id":"openstack%2Fcharm-nova-compute~master~Id2db3a70b8d1287bda006f1bbc5442038f7070f1","project":"openstack/charm-nova-compute","branch":"master","attention_set":{},"removed_from_attention_set":{"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2022-06-27 13:25:31.000000000","reason":"\u003cGERRIT_ACCOUNT_10058\u003e replied on the change","reason_account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2022-06-29 05:27:36.000000000","reason":"Change was submitted"},"10366":{"account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"last_update":"2022-06-28 02:55:24.000000000","reason":"\u003cGERRIT_ACCOUNT_10366\u003e replied on the change","reason_account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}}},"hashtags":[],"change_id":"Id2db3a70b8d1287bda006f1bbc5442038f7070f1","subject":"AppArmor policy update for os-brick and iSCSI","status":"MERGED","created":"2022-06-24 14:26:49.000000000","updated":"2022-06-29 05:27:36.000000000","submitted":"2022-06-29 05:27:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"847582","meta_rev_id":"37da0cdcf8f88a485be39a7853783675c5ee46e6","_number":847582,"virtual_id_number":847582,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-06-29 03:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 05:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"recommended":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-27 13:25:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-28 02:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-29 05:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 14:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 15:21:05.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 17:35:41.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"CC"},{"updated":"2022-06-24 17:35:52.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2022-06-27 04:25:20.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"}],"messages":[{"id":"a13dc60a6750dc236ee785030b556abcf53f4aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-24 14:26:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10b2b6b0361b04f9e77eb1fd0b9e0061d8632e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-24 14:28:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88f555dd482b47c1225307bddf0c45999d12093b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 14:45:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad70fa549f14cf291c44c1dd37a99fd : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01b81edbb84f4c73bae4512c059adf6a : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a3e794fde204af0843c11c1927e4806 : FAILURE in 7m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f6c9967a47c49289fd7e5dffe24979a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fd055257733f294e817d69c383fa82c619f992e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-24 14:59:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"976c53fd8e348b0d1a221e39433d8dd388c02f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 15:13:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636e8a0e3e1d4be8b79d1ed556b6796c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca360d5e26c410381a7f115833ab1ef : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94c6dfc2d2a844aaabea050770a48f60 : FAILURE in 8m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5271493f9be443fcb54f656c9ebacbab : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"92a22c35f86496adfa87b251a83fcefa4ab066b6","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-24 15:21:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/128/847582/3/check/charm-build/128d18f/ : SUCCESS in 9m 49s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/dbb/847582/3/check/osci-lint/dbbc91d/ : SUCCESS in 8m 56s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/d77/847582/3/check/tox-py38/d776915/ : SUCCESS in 10m 22s\n- tox-py310 https://openstack-ci-reports.ubuntu.com/artifacts/7c7/847582/3/check/tox-py310/7c7cdf5/ : RETRY_LIMIT in 5m 09s\n- ceph-focal-xena_nova ceph-focal-xena_nova : SKIPPED\n- ceph-focal-yoga_nova ceph-focal-yoga_nova : SKIPPED\n- ceph-jammy-yoga_nova ceph-jammy-yoga_nova : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd5882b1c942df0d44874ac92c05a7c3803aabc3","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2022-06-24 17:35:41.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"97f1c4138117154616f1a08a05871483a25500db","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2022-06-24 17:35:52.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"2dc54326ee3bc4a393136813b722e62e1654c369","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-25 00:32:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9bf429c5c0c999c017eaa9ae31b7878588b6843d","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-25 02:26:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a4c3f9a2dcdfa5aaa5955c4cf79b585d289232c","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-25 02:38:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1fcf3b5e580d733e3ea3d6683e4203228b76f262","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-25 03:00:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1785f45dd060f32b0951275d37ed6dd40eaeff01","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-25 03:22:44.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/027/847582/3/check/charm-build/027cf48/ : SUCCESS in 9m 47s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/13e/847582/3/check/osci-lint/13e8936/ : SUCCESS in 10m 17s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/ee7/847582/3/check/tox-py38/ee70d57/ : SUCCESS in 9m 19s\n- tox-py310 https://openstack-ci-reports.ubuntu.com/artifacts/ee0/847582/3/check/tox-py310/ee04ba6/ : RETRY_LIMIT in 4m 35s\n- ceph-focal-xena_nova ceph-focal-xena_nova : SKIPPED\n- ceph-focal-yoga_nova ceph-focal-yoga_nova : SKIPPED\n- ceph-jammy-yoga_nova ceph-jammy-yoga_nova : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5cc8c1017519f2251095e85140c1357f36ce6f9a","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-25 03:25:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49771e956dba414fa1c39378d8355f184bc2a47c","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-25 03:56:08.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/6c2/847582/3/check/charm-build/6c2983a/ : SUCCESS in 9m 26s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/94e/847582/3/check/osci-lint/94e80ed/ : SUCCESS in 8m 45s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/f17/847582/3/check/tox-py38/f17c2eb/ : SUCCESS in 11m 27s\n- tox-py310 https://openstack-ci-reports.ubuntu.com/artifacts/b54/847582/3/check/tox-py310/b540c7d/ : RETRY_LIMIT in 7m 16s\n- ceph-focal-xena_nova ceph-focal-xena_nova : SKIPPED\n- ceph-focal-yoga_nova ceph-focal-yoga_nova : SKIPPED\n- ceph-jammy-yoga_nova ceph-jammy-yoga_nova : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a024b8ca0b2f38b0c9a50f3cc6afc6fc0bd657a8","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2022-06-27 13:25:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8e376f5860ac87446373a05ee8e1d58ad0f61ac6","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-27 13:30:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fa00f02d98a54448acbcef640d2dbda227dcb34","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-27 15:18:15.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/623/847582/3/check/charm-build/6237c7c/ : SUCCESS in 9m 35s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/69e/847582/3/check/osci-lint/69e891c/ : SUCCESS in 8m 54s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/62a/847582/3/check/tox-py38/62a1da5/ : SUCCESS in 10m 50s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/b11/847582/3/check/ceph-focal-xena_nova/b117a19/ : FAILURE in 1h 34m 57s\n- ceph-focal-yoga_nova ceph-focal-yoga_nova : SKIPPED\n- ceph-jammy-yoga_nova ceph-jammy-yoga_nova : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ef22f185f4806b2af3df67aef00eb1c04b55aba","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-28 00:26:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b0ebb669617ebee45977d059ca22c06d1d76c2c7","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-06-28 02:55:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3030ef42d82f1bb2b8bb5692e7a30eb8962ccc2a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-28 04:23:57.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/f7e/847582/3/check/charm-build/f7e872e/ : SUCCESS in 8m 17s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/aa2/847582/3/check/osci-lint/aa2d54c/ : SUCCESS in 8m 41s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/f13/847582/3/check/tox-py38/f138293/ : SUCCESS in 8m 34s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/733/847582/3/check/ceph-focal-xena_nova/733370e/ : SUCCESS in 1h 35m 53s\n- ceph-focal-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/366/847582/3/check/ceph-focal-yoga_nova/366cfcd/ : FAILURE in 1h 35m 15s\n- ceph-jammy-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/b8b/847582/3/check/ceph-jammy-yoga_nova/b8b6c1d/ : FAILURE in 1h 50m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c45b165b8640758933db14ef6ab4e0620d71568","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-28 04:27:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd02ddf50aaa19a596b88634d7a66a60f205f70c","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-28 04:28:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4304c0d5ed0c0124a248b06b4bc242609a6bf433","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-28 05:28:59.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/280/847582/3/check/charm-build/2807b07/ : SUCCESS in 9m 17s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/6cc/847582/3/check/osci-lint/6ccd29e/ : SUCCESS in 9m 15s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/b3c/847582/3/check/tox-py38/b3c9a58/ : SUCCESS in 12m 19s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/77d/847582/3/check/ceph-focal-xena_nova/77d3688/ : FAILURE in 47m 14s\n- ceph-focal-yoga_nova ceph-focal-yoga_nova : SKIPPED\n- ceph-jammy-yoga_nova ceph-jammy-yoga_nova : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65fc3ee5fc57a60faf06749bd6722f0af3149eab","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-28 06:40:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"231a3fc91ce0e83f9e99755c83d1f09feea59e4b","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-28 07:35:01.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/7fe/847582/3/check/charm-build/7fe3a4c/ : FAILURE in 53m 02s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/ae8/847582/3/check/osci-lint/ae87094/ : SUCCESS in 7m 53s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/1b4/847582/3/check/tox-py38/1b492ea/ : SUCCESS in 12m 15s\n- ceph-focal-xena_nova ceph-focal-xena_nova : SKIPPED\n- ceph-focal-yoga_nova ceph-focal-yoga_nova : SKIPPED\n- ceph-jammy-yoga_nova ceph-jammy-yoga_nova : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b583da7062919926f009cc7d672e555368d63768","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2022-06-28 11:24:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b12adc04dd833c3a3919a5498b920178a1cce464","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-28 15:31:43.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/4c1/847582/3/check/charm-build/4c11ea0/ : SUCCESS in 9m 51s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/8e5/847582/3/check/osci-lint/8e55adf/ : SUCCESS in 8m 52s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/41c/847582/3/check/tox-py38/41c127b/ : SUCCESS in 10m 34s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/3ad/847582/3/check/ceph-focal-xena_nova/3ad4153/ : SUCCESS in 1h 51m 38s\n- ceph-focal-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/e3c/847582/3/check/ceph-focal-yoga_nova/e3cb753/ : FAILURE in 1h 12m 56s\n- ceph-jammy-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/850/847582/3/check/ceph-jammy-yoga_nova/850ac3e/ : SUCCESS in 2h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c1a0463a0756b2dd50ff0d6da2559042e4db776","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-28 23:27:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5aafbfe7f2a5676b659c361b5fa8836af1ad5b67","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-29 03:44:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/c06/847582/3/check/charm-build/c063878/ : SUCCESS in 9m 56s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/1ae/847582/3/check/osci-lint/1ae85cc/ : SUCCESS in 9m 40s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/112/847582/3/check/tox-py38/112ce1a/ : SUCCESS in 11m 56s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/836/847582/3/check/ceph-focal-xena_nova/836b739/ : SUCCESS in 1h 42m 05s\n- ceph-focal-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/ef3/847582/3/check/ceph-focal-yoga_nova/ef383a0/ : SUCCESS in 1h 45m 03s\n- ceph-jammy-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/925/847582/3/check/ceph-jammy-yoga_nova/925d7e5/ : SUCCESS in 1h 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a0c9af359b5cca797092101458234fd922e9dc6a","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-06-29 05:17:33.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc3fc61f2886a4eff5b0398a3e2c3f24c758a6a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:17:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"074baa960459c45697198c07dbef8e48914889af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:27:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1d4691d9304d8eb61b34b865980484 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d36db39a806548aebb14ad34a18c8a74 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"37da0cdcf8f88a485be39a7853783675c5ee46e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:27:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cf0f464391df509e752c6010964efe2aca10ef89","revisions":{"42b696f8a0a9c630ac13062e00d6f6a335a447b8":{"kind":"REWORK","_number":1,"created":"2022-06-24 14:26:49.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/82/847582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/82/847582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/1"}}},"commit":{"parents":[{"commit":"06856f151cb4d7f7c1844513edfe32f6f3ff58ed","subject":"Ensure vgpu services restarted in package_upgrade action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/06856f151cb4d7f7c1844513edfe32f6f3ff58ed"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"subject":"AppArmor policy update for os-brick lock files","message":"AppArmor policy update for os-brick lock files\n\nIn iSCSI usecase including cinder-lvm, os-brick requires lock files such\nas:\n- /run/lock/nova/os-brick-connect_volume\n- /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nCloses-Bug: #1979812\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/42b696f8a0a9c630ac13062e00d6f6a335a447b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/42b696f8a0a9c630ac13062e00d6f6a335a447b8"}]},"branch":"refs/heads/master"},"8a3896d1c60c6dcc52ba680e968e7ce66d266ea1":{"kind":"REWORK","_number":2,"created":"2022-06-24 14:28:23.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/82/847582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/82/847582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/2"}}},"commit":{"parents":[{"commit":"06856f151cb4d7f7c1844513edfe32f6f3ff58ed","subject":"Ensure vgpu services restarted in package_upgrade action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/06856f151cb4d7f7c1844513edfe32f6f3ff58ed"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:27:57.000000000","tz":540},"subject":"AppArmor policy update for os-brick lock files","message":"AppArmor policy update for os-brick lock files\n\nIn iSCSI usecase including cinder-lvm, os-brick requires lock files such\nas:\n- /run/lock/nova/os-brick-connect_volume\n- /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nCloses-Bug: #1979812\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/8a3896d1c60c6dcc52ba680e968e7ce66d266ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/8a3896d1c60c6dcc52ba680e968e7ce66d266ea1"}]},"branch":"refs/heads/master"},"cf0f464391df509e752c6010964efe2aca10ef89":{"kind":"REWORK","_number":3,"created":"2022-06-24 14:59:45.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/82/847582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/82/847582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/82/847582/3"}}},"commit":{"parents":[{"commit":"06856f151cb4d7f7c1844513edfe32f6f3ff58ed","subject":"Ensure vgpu services restarted in package_upgrade action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/06856f151cb4d7f7c1844513edfe32f6f3ff58ed"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:58:32.000000000","tz":540},"subject":"AppArmor policy update for os-brick and iSCSI","message":"AppArmor policy update for os-brick and iSCSI\n\nIn iSCSI usecases including cinder-lvm, os-brick requires lock files\nsuch as:\n  - /run/lock/nova/os-brick-connect_volume\n  - /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nand lsscsi requires following access to compose a rescan command such as\n\"/sys/bus/scsi/drivers/sd/2:0:0:0/rescan\":\n  - /dev/\n  - /sys/bus/scsi/devices/\n\nCloses-Bug: #1979812\nRelated-Bug: #1939390\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/cf0f464391df509e752c6010964efe2aca10ef89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/cf0f464391df509e752c6010964efe2aca10ef89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}}]}],"submit_requirements":[]}
