)]}'
{"id":"openstack%2Fcharm-nova-compute~848055","triplet_id":"openstack%2Fcharm-nova-compute~stable%2Fxena~Id2db3a70b8d1287bda006f1bbc5442038f7070f1","project":"openstack/charm-nova-compute","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2022-08-24 09:49:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2db3a70b8d1287bda006f1bbc5442038f7070f1","subject":"AppArmor policy update for os-brick and iSCSI","status":"MERGED","created":"2022-06-29 05:41:31.000000000","updated":"2022-08-24 09:49:56.000000000","submitted":"2022-08-24 09:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848055","cherry_pick_of_change":847582,"cherry_pick_of_patch_set":3,"meta_rev_id":"b9a1cdada0f7b82a8d6945abbcca2a6c4a075043","_number":848055,"virtual_id_number":848055,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2022-07-21 03:36:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 09:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":2,"date":"2022-08-24 09:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-24 08:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":1,"date":"2022-08-24 09:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 05:52:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 06:34:02.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-01 02:43:09.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"CC"},{"updated":"2022-08-24 08:14:52.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2022-08-24 09:43:25.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"}],"messages":[{"id":"5caa3616998dd9d2a5a495347024ce5012bda356","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-29 05:41:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1e69a530b63aaf752b25a0dfd34f0c8158b7e6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:52:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c87b53afc2d4377b45b3128752da364 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d1be017acc44fffb1bbfc4be407440a : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75e7e94d9c6a47c093167ed077267652 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d965cd43b3bc43119cdd66ba39c19ff4 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b554b24d2d34c35b9410cbe265e3550 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6897e8890d1279596aa8b111ec2d4ae396959ea1","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-29 06:34:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/a03/848055/1/check/charm-build/a03dee7/ : SUCCESS in 10m 35s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/14b/848055/1/check/osci-lint/14b25ad/ : SUCCESS in 9m 32s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/816/848055/1/check/tox-py38/816f827/ : SUCCESS in 11m 23s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/1de/848055/1/check/ceph-focal-xena_nova/1ded1c1/ : FAILURE in 38m 40s\n- ceph-focal-wallaby_nova ceph-focal-wallaby_nova : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"d66dbd4b01e5421be28c4d78edf8c71bfb0c786a","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-07-01 02:43:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de5dd5c96e706d5b7bf2654ab80463339ce1dab6","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-01 04:38:20.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/e64/848055/1/check/charm-build/e646120/ : SUCCESS in 10m 53s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/ff5/848055/1/check/osci-lint/ff52fb0/ : SUCCESS in 8m 04s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/416/848055/1/check/tox-py38/416a2e1/ : SUCCESS in 12m 43s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/396/848055/1/check/ceph-focal-xena_nova/396c4c3/ : FAILURE in 43m 00s\n- ceph-focal-wallaby_nova https://openstack-ci-reports.ubuntu.com/artifacts/2c2/848055/1/check/ceph-focal-wallaby_nova/2c27864/ : FAILURE in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"45d2df445867fa4b10ba22bc6a0357a27b891ef1","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-07-06 13:38:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e24572760ca04fc6e1c549d475bcf21f0643e66b","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-06 20:48:23.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/919/848055/1/check/charm-build/919f352/ : SUCCESS in 19m 04s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/104/848055/1/check/osci-lint/1043446/ : SUCCESS in 11m 07s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/bba/848055/1/check/tox-py38/bbad78d/ : SUCCESS in 9m 41s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/084/848055/1/check/ceph-focal-xena_nova/084cbff/ : FAILURE in 2h 30m 42s\n- ceph-focal-wallaby_nova https://openstack-ci-reports.ubuntu.com/artifacts/210/848055/1/check/ceph-focal-wallaby_nova/210e7db/ : SUCCESS in 1h 53m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c4ab5619674c44ea1c7c485e86c6396563b47b04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-07-21 02:05:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a7957047fc7e368a6347ced13e5dcd37cae2b862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 02:35:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/838b62ed1a7a4391b9a241e9ec50c863\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af609653280746b6b283e5f8610f917c : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87e860b10b8c42fd8eb209b0a6e46fe5 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1fc5dcd3e944e090603403bf6ae1de : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ba93e863b20466ab6f4ed20ab60fdc7 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bc9acae3952492981ad970e2ba67bcd : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6280de07c1a383c0cdf6c82249d9737ff606bed0","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-21 03:36:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/8c3/848055/2/check/charm-build/8c3593e/ : SUCCESS in 10m 06s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/42a/848055/2/check/osci-lint/42a1245/ : SUCCESS in 6m 47s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/bac/848055/2/check/tox-py38/bac815a/ : SUCCESS in 8m 09s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/32d/848055/2/check/ceph-focal-xena_nova/32dd415/ : SUCCESS in 1h 15m 19s\n- ceph-focal-wallaby_nova https://openstack-ci-reports.ubuntu.com/artifacts/76c/848055/2/check/ceph-focal-wallaby_nova/76cbd9e/ : SUCCESS in 1h 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"bd2274931e26ef3cc87f215262863b5823887615","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-08-24 08:14:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41add114c0c274e5bb029f68758948578af97599","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-08-24 09:43:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03a7bee94161218c0da9fd808da60e7c6571d5e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:43:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4a8cf803075b663913dfb5e9b4e6c15a162633c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:49:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/092c61f4e10d4b93bbc6a6c78a069545\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d24111039a5643c2983f1693502d5194 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b317b1c94b8448b7924bdf1a4c4701c2 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a87f1f8a5ca4c8198c02519e2a386d5 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b9a1cdada0f7b82a8d6945abbcca2a6c4a075043","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:49:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c16a9aedab7eed0699f3d677eab03374d348e6a8","revisions":{"12c97d9de0c07819c8cdf136ea94202df753ad48":{"kind":"REWORK","_number":1,"created":"2022-06-29 05:41:31.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/55/848055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/55/848055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/1"}}},"commit":{"parents":[{"commit":"f3707dc330e101fd1d3eaaffe05d7e71d7d69daf","subject":"Update the charmcraft.yaml for xena (no 22.04 base)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/f3707dc330e101fd1d3eaaffe05d7e71d7d69daf"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-29 05:41:31.000000000","tz":0},"subject":"AppArmor policy update for os-brick and iSCSI","message":"AppArmor policy update for os-brick and iSCSI\n\nIn iSCSI usecases including cinder-lvm, os-brick requires lock files\nsuch as:\n  - /run/lock/nova/os-brick-connect_volume\n  - /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nand lsscsi requires following access to compose a rescan command such as\n\"/sys/bus/scsi/drivers/sd/2:0:0:0/rescan\":\n  - /dev/\n  - /sys/bus/scsi/devices/\n\nCloses-Bug: #1979812\nRelated-Bug: #1939390\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n(cherry picked from commit cf0f464391df509e752c6010964efe2aca10ef89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/12c97d9de0c07819c8cdf136ea94202df753ad48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/12c97d9de0c07819c8cdf136ea94202df753ad48"}]},"branch":"refs/heads/stable/xena"},"c16a9aedab7eed0699f3d677eab03374d348e6a8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-21 02:05:22.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/55/848055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/55/848055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/848055/2"}}},"commit":{"parents":[{"commit":"64d8b2e142e3098d56ac30cf2ca9c70c65adfd00","subject":"Fix invalid tempest pin for 31.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/64d8b2e142e3098d56ac30cf2ca9c70c65adfd00"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-07-21 02:05:22.000000000","tz":0},"subject":"AppArmor policy update for os-brick and iSCSI","message":"AppArmor policy update for os-brick and iSCSI\n\nIn iSCSI usecases including cinder-lvm, os-brick requires lock files\nsuch as:\n  - /run/lock/nova/os-brick-connect_volume\n  - /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nand lsscsi requires following access to compose a rescan command such as\n\"/sys/bus/scsi/drivers/sd/2:0:0:0/rescan\":\n  - /dev/\n  - /sys/bus/scsi/devices/\n\nCloses-Bug: #1979812\nRelated-Bug: #1939390\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n(cherry picked from commit cf0f464391df509e752c6010964efe2aca10ef89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/c16a9aedab7eed0699f3d677eab03374d348e6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/c16a9aedab7eed0699f3d677eab03374d348e6a8"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}}]}],"submit_requirements":[]}
