)]}'
{"id":"openstack%2Fcharm-nova-compute~848056","triplet_id":"openstack%2Fcharm-nova-compute~stable%2Fwallaby~Id2db3a70b8d1287bda006f1bbc5442038f7070f1","project":"openstack/charm-nova-compute","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2022-08-24 09:49:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2db3a70b8d1287bda006f1bbc5442038f7070f1","subject":"AppArmor policy update for os-brick and iSCSI","status":"MERGED","created":"2022-06-29 05:41:50.000000000","updated":"2022-08-24 09:49:40.000000000","submitted":"2022-08-24 09:49:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848056","cherry_pick_of_change":847582,"cherry_pick_of_patch_set":3,"meta_rev_id":"6255a85f67913b6c8f1946d041b4c98421a3e72b","_number":848056,"virtual_id_number":848056,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2022-07-29 16:26:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 09:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":2,"date":"2022-08-24 09:43:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-24 08:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":1,"date":"2022-08-24 09:43:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 05:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 06:36:50.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-01 02:43:17.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"CC"},{"updated":"2022-08-24 08:15:28.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2022-08-24 09:43:01.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"}],"messages":[{"id":"7c54674a07391982cd96fb71d673c31643e05863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-29 05:41:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"890d25b6bb18c9dd734340cce727f20857251734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:53:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/529df08b36bd4585aee67d90bb5f1a04 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08dcf145784243dc98aa0f9a0470e276 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75bf092407904b32859c11ff0b0e70a3 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8701d84b944742f1a3332f11381d7773 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2bc8cde97004a3ca8e5032a26475919 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"77431028b4012c39e876bb1b82b0788b79fd27e7","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-29 06:36:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/c81/848056/1/check/charm-build/c812c68/ : SUCCESS in 10m 50s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/ca0/848056/1/check/osci-lint/ca0fe32/ : SUCCESS in 11m 17s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/886/848056/1/check/tox-py38/88611e6/ : SUCCESS in 12m 12s\n- ceph-focal-wallaby_nova https://openstack-ci-reports.ubuntu.com/artifacts/803/848056/1/check/ceph-focal-wallaby_nova/8037f61/ : FAILURE in 41m 25s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/4ff/848056/1/check/ceph-focal-victoria_nova/4ff2bfd/ : FAILURE in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9b9babd71d8a4b21b8c4b182bf0f4894cdcf8d94","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-07-01 02:43:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"243a3dd94e266f80c322486bf51e7372c9ea490a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-01 04:19:55.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/b1c/848056/1/check/charm-build/b1ca2be/ : SUCCESS in 10m 10s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/1ee/848056/1/check/osci-lint/1ee3836/ : SUCCESS in 12m 12s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/c0a/848056/1/check/tox-py38/c0a30cf/ : SUCCESS in 10m 44s\n- ceph-focal-wallaby_nova https://openstack-ci-reports.ubuntu.com/artifacts/4c7/848056/1/check/ceph-focal-wallaby_nova/4c73058/ : FAILURE in 1h 20m 08s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/077/848056/1/check/ceph-focal-victoria_nova/077bf1a/ : FAILURE in 1h 22m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7c9d56b17e47912ddff21c6144768f305b5dae4f","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-07-06 13:40:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87ec94ee5a30291e73601978ab9ef108c728d5b7","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-06 19:08:48.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/3e9/848056/1/check/charm-build/3e915b3/ : SUCCESS in 14m 52s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/393/848056/1/check/osci-lint/3937376/ : SUCCESS in 8m 22s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/c63/848056/1/check/tox-py38/c634322/ : SUCCESS in 14m 37s\n- ceph-focal-wallaby_nova https://openstack-ci-reports.ubuntu.com/artifacts/2ce/848056/1/check/ceph-focal-wallaby_nova/2ceef2d/ : FAILURE in 48m 51s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/4bf/848056/1/check/ceph-focal-victoria_nova/4bf4d78/ : FAILURE in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"449c9173663064a1ca0019dffc794682bfacadcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-07-29 14:45:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b6548dbf273f2d7f8a9784695ca3382cf023c0d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:55:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd547ec202704686a6ae37d1422ee0be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76453d34e2a1451fa7c9178db5935166 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34d3933b01cf4b95965ed47b12c7959b : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e32f85bae5be423aae1dd1b07a2da7dc : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23db8106c7ee44f881f4589d949d58c7 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efb179d69ebf4bcfac9efb1c15ed7509 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9740c0a6926ede231e44c3f6fa29fc4c7842c8ce","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-29 16:26:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/a27/848056/2/check/charm-build/a27c149/ : SUCCESS in 8m 56s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/c78/848056/2/check/osci-lint/c78225b/ : SUCCESS in 8m 06s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/c6f/848056/2/check/tox-py38/c6f5970/ : SUCCESS in 9m 39s\n- ceph-focal-wallaby_nova https://openstack-ci-reports.ubuntu.com/artifacts/e71/848056/2/check/ceph-focal-wallaby_nova/e717df4/ : SUCCESS in 1h 22m 30s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/30b/848056/2/check/ceph-focal-victoria_nova/30be1a4/ : SUCCESS in 1h 24m 16s","accounts_in_message":[],"_revision_number":2},{"id":"91d6399f617cd3ebbb5e7b008a2127f704f5961c","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-08-24 08:15:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92ee76834c728b20d1cc34fa3f3ed6a20d06c6d4","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-08-24 09:43:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9a8d54130c680e9d2f10539fc7c104a19098058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:43:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"834e2eb75c323b252086604b355972bc8f5c0eff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:49:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b67cbf76a1b44a79b2b97e4d65930fec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54abd2ca7321479fb93ff3f72abb2ece : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4805bb75ce4f219e9543926ab968ed : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e8f92a2e12a4fb593f73edc4583c0d7 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"6255a85f67913b6c8f1946d041b4c98421a3e72b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:49:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7fb373d9fa5694f9d53aac261c2650224f8af99e","revisions":{"a5fddc93836ca4603c1c841839144815d0020d45":{"kind":"REWORK","_number":1,"created":"2022-06-29 05:41:50.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/56/848056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/56/848056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/1"}}},"commit":{"parents":[{"commit":"fd08a92b526934842291a834d65216efacca7fac","subject":"Pin tempest for python\u003c3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/fd08a92b526934842291a834d65216efacca7fac"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-29 05:41:50.000000000","tz":0},"subject":"AppArmor policy update for os-brick and iSCSI","message":"AppArmor policy update for os-brick and iSCSI\n\nIn iSCSI usecases including cinder-lvm, os-brick requires lock files\nsuch as:\n  - /run/lock/nova/os-brick-connect_volume\n  - /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nand lsscsi requires following access to compose a rescan command such as\n\"/sys/bus/scsi/drivers/sd/2:0:0:0/rescan\":\n  - /dev/\n  - /sys/bus/scsi/devices/\n\nCloses-Bug: #1979812\nRelated-Bug: #1939390\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n(cherry picked from commit cf0f464391df509e752c6010964efe2aca10ef89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/a5fddc93836ca4603c1c841839144815d0020d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/a5fddc93836ca4603c1c841839144815d0020d45"}]},"branch":"refs/heads/stable/wallaby"},"7fb373d9fa5694f9d53aac261c2650224f8af99e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-29 14:45:31.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/56/848056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/56/848056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/56/848056/2"}}},"commit":{"parents":[{"commit":"4e69539db5ee7b8fb30212b34612e981b6580db3","subject":"Merge \"Update the bundles to use the stable track, edge risk for wallaby\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/4e69539db5ee7b8fb30212b34612e981b6580db3"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-07-29 14:45:31.000000000","tz":0},"subject":"AppArmor policy update for os-brick and iSCSI","message":"AppArmor policy update for os-brick and iSCSI\n\nIn iSCSI usecases including cinder-lvm, os-brick requires lock files\nsuch as:\n  - /run/lock/nova/os-brick-connect_volume\n  - /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nand lsscsi requires following access to compose a rescan command such as\n\"/sys/bus/scsi/drivers/sd/2:0:0:0/rescan\":\n  - /dev/\n  - /sys/bus/scsi/devices/\n\nCloses-Bug: #1979812\nRelated-Bug: #1939390\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n(cherry picked from commit cf0f464391df509e752c6010964efe2aca10ef89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/7fb373d9fa5694f9d53aac261c2650224f8af99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/7fb373d9fa5694f9d53aac261c2650224f8af99e"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}}]}],"submit_requirements":[]}
