)]}'
{"id":"openstack%2Fcharm-nova-compute~848057","triplet_id":"openstack%2Fcharm-nova-compute~stable%2Fvictoria~Id2db3a70b8d1287bda006f1bbc5442038f7070f1","project":"openstack/charm-nova-compute","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2022-08-25 04:01:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2db3a70b8d1287bda006f1bbc5442038f7070f1","subject":"AppArmor policy update for os-brick and iSCSI","status":"MERGED","created":"2022-06-29 05:42:09.000000000","updated":"2022-08-25 04:01:02.000000000","submitted":"2022-08-25 04:01:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848057","cherry_pick_of_change":847582,"cherry_pick_of_patch_set":3,"meta_rev_id":"c3889001ace50befcc3eaa4b225d7b481e893ae1","_number":848057,"virtual_id_number":848057,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2022-08-24 18:29:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-25 04:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":2,"date":"2022-08-24 21:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-25 03:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-25 03:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 05:54:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 06:31:55.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-01 02:43:23.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"CC"},{"updated":"2022-08-24 16:56:43.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"CC"},{"updated":"2022-08-24 21:50:42.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2022-08-25 03:55:32.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"}],"messages":[{"id":"0e9141df9669f8d79e74ef9be45631b2522ab452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-06-29 05:42:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7dedc345c383d2dcd66ff528371642c04b0d4546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:54:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0336d0a02f954343ab160498cda451df : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5061db368d4a4b44ba4e3f42d6162544 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1e01acc735c4d9aa7507b6e5314b32e : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ffd39564ec345489199f52cb5cf08a0 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cc035fccc984692fb4b87e53b74dfac66308939e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-29 06:31:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/869/848057/1/check/charm-build/869b47d/ : SUCCESS in 9m 07s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/251/848057/1/check/osci-lint/251f553/ : SUCCESS in 6m 44s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/be7/848057/1/check/tox-py38/be7d80e/ : SUCCESS in 10m 48s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/861/848057/1/check/ceph-focal-victoria_nova/861de1d/ : FAILURE in 29m 53s\n- ceph-focal-ussuri_nova https://openstack-ci-reports.ubuntu.com/artifacts/277/848057/1/check/ceph-focal-ussuri_nova/2772448/ : FAILURE in 26m 39s","accounts_in_message":[],"_revision_number":1},{"id":"124b29928837d069daaa6fb55608e6249d6cd4b2","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-07-01 02:43:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9365d7f64ac91d225c1c8410169538ee97c6ee27","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-01 05:05:09.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/4a7/848057/1/check/charm-build/4a71946/ : SUCCESS in 10m 13s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/217/848057/1/check/osci-lint/217b60a/ : SUCCESS in 8m 11s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/438/848057/1/check/tox-py38/438a8c5/ : SUCCESS in 13m 08s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/8a5/848057/1/check/ceph-focal-victoria_nova/8a5011d/ : FAILURE in 1h 23m 49s\n- ceph-focal-ussuri_nova https://openstack-ci-reports.ubuntu.com/artifacts/299/848057/1/check/ceph-focal-ussuri_nova/299c4fa/ : SUCCESS in 1h 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c739f4a9eb3d4d3ce6b812f40817cb33713c0a64","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-07-06 13:36:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7703989886e17a2061a7901b166a135fb0514891","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-06 18:28:21.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/f45/848057/1/check/charm-build/f450be6/ : SUCCESS in 16m 55s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/f9d/848057/1/check/osci-lint/f9dcaa2/ : SUCCESS in 16m 47s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/f26/848057/1/check/tox-py38/f26fdb3/ : SUCCESS in 16m 26s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/a34/848057/1/check/ceph-focal-victoria_nova/a34508b/ : FAILURE in 1h 01m 06s\n- ceph-focal-ussuri_nova https://openstack-ci-reports.ubuntu.com/artifacts/ad7/848057/1/check/ceph-focal-ussuri_nova/ad7bdef/ : FAILURE in 1h 02m 53s","accounts_in_message":[],"_revision_number":1},{"id":"70cfb917e1be344740d45e46909cbdfa69638ec7","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2022-08-24 08:12:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b99beadc4a5c90ed97ce8451d45674373096365f","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-24 12:43:40.000000000","message":"Patch Set 1:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/fc0/848057/1/check/charm-build/fc07f0e/ : SUCCESS in 8m 50s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/56f/848057/1/check/osci-lint/56f954e/ : SUCCESS in 7m 28s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/372/848057/1/check/tox-py38/3720e18/ : SUCCESS in 9m 24s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/fae/848057/1/check/ceph-focal-victoria_nova/fae5316/ : TIMED_OUT in 4h 14m 14s\n- ceph-focal-ussuri_nova https://openstack-ci-reports.ubuntu.com/artifacts/0b7/848057/1/check/ceph-focal-ussuri_nova/0b77810/ : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5afe73e1340c8f500364b27c00c745e96c63fa9e","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-08-24 16:56:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c086863ec3fee668a375ecbc7e008f4abd04fefb","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-24 18:29:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/83b/848057/1/check/charm-build/83bbf44/ : SUCCESS in 8m 30s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/d19/848057/1/check/osci-lint/d19b108/ : SUCCESS in 8m 26s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/d9a/848057/1/check/tox-py38/d9acb33/ : SUCCESS in 8m 10s\n- ceph-focal-victoria_nova https://openstack-ci-reports.ubuntu.com/artifacts/426/848057/1/check/ceph-focal-victoria_nova/4268dfb/ : SUCCESS in 1h 22m 16s\n- ceph-focal-ussuri_nova https://openstack-ci-reports.ubuntu.com/artifacts/146/848057/1/check/ceph-focal-ussuri_nova/146f179/ : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"968e0b462efe4e586bd9ef4535e52d953d4f3685","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-08-24 21:50:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9653e1d2aaf3d6eda408ba777fac1147a5c85814","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-08-25 03:55:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af7875001ae99a7c310d3134680eb744a922d388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 03:55:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8636131a751e8c93129e7ef906d034875eb22588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 04:01:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ce61f057ac411c8816a8afcb9eded8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a06d5b0f3ad45049acf67b67eceebbb : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c1d382b754544dc9d226ad39981998d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cddd7e8932f54d90b9532f06395a0675 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c3889001ace50befcc3eaa4b225d7b481e893ae1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 04:01:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5e658d04571a7c3019b5c5b711aca632989248c","revisions":{"b5e658d04571a7c3019b5c5b711aca632989248c":{"kind":"REWORK","_number":1,"created":"2022-06-29 05:42:09.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/57/848057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/57/848057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/57/848057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/57/848057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/57/848057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/57/848057/1"}}},"commit":{"parents":[{"commit":"28f0b70bce3a53c6f01ff8a974465e4c0b5b6997","subject":"Make sure lsscsi is installed for os-brick","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/28f0b70bce3a53c6f01ff8a974465e4c0b5b6997"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-24 14:22:54.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2022-06-29 05:42:09.000000000","tz":0},"subject":"AppArmor policy update for os-brick and iSCSI","message":"AppArmor policy update for os-brick and iSCSI\n\nIn iSCSI usecases including cinder-lvm, os-brick requires lock files\nsuch as:\n  - /run/lock/nova/os-brick-connect_volume\n  - /run/lock/nova/os-brick-connect_to_iscsi_portal-192.168.0.1\n\nand lsscsi requires following access to compose a rescan command such as\n\"/sys/bus/scsi/drivers/sd/2:0:0:0/rescan\":\n  - /dev/\n  - /sys/bus/scsi/devices/\n\nCloses-Bug: #1979812\nRelated-Bug: #1939390\nChange-Id: Id2db3a70b8d1287bda006f1bbc5442038f7070f1\n(cherry picked from commit cf0f464391df509e752c6010964efe2aca10ef89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/b5e658d04571a7c3019b5c5b711aca632989248c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/b5e658d04571a7c3019b5c5b711aca632989248c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}}]}],"submit_requirements":[]}
