)]}'
{"id":"openstack%2Fcharm-nova-compute~870050","triplet_id":"openstack%2Fcharm-nova-compute~stable%2Fyoga~I4193c042f6b3a55dfb1dc57f0f4d3bc71e19006f","project":"openstack/charm-nova-compute","branch":"stable/yoga","topic":"bug/1998300-stable/yoga","attention_set":{},"removed_from_attention_set":{"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2023-01-18 03:08:06.000000000","reason":"Change was submitted"},"10366":{"account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"last_update":"2023-01-18 03:08:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4193c042f6b3a55dfb1dc57f0f4d3bc71e19006f","subject":"Use enabled option for vnc","status":"MERGED","created":"2023-01-13 02:31:34.000000000","updated":"2023-01-18 03:08:06.000000000","submitted":"2023-01-18 03:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870050-bug/1998300-stable/yoga","cherry_pick_of_change":866246,"cherry_pick_of_patch_set":2,"meta_rev_id":"eefa7dd535d5ede05e20ef2c5e1901bbfabf959e","_number":870050,"virtual_id_number":870050,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":1,"date":"2023-01-17 06:17:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 03:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":2,"date":"2023-01-18 02:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":2,"date":"2023-01-17 08:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-18 02:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":1,"date":"2023-01-18 02:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-18 02:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-13 02:41:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-13 04:21:34.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 08:27:22.000000000","updated_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"reviewer":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"state":"REVIEWER"},{"updated":"2023-01-18 02:58:08.000000000","updated_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"reviewer":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"state":"REVIEWER"},{"updated":"2023-01-18 02:58:09.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"}],"messages":[{"id":"0b6329d04252cc00863d73dfe8d9610880b76973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2023-01-13 02:31:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"518281e7bd9333cd028ee5292e0eb8757a47dd94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 02:41:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed45c26c152f4629a8997f5e4009dfd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47af39242a6d49e5a6223541de9cae11 : FAILURE in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3a604277c3a47adbeb6c819178009a2 : FAILURE in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06b7d7799e9246218b4794fe50b23cdd : FAILURE in 2m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ddf09f3b9854a2e83c2453e92c66d02 : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"525a01afba158d16f34b4ad1a3356cd47637e57a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-01-13 04:21:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/93f/870050/1/check/charm-build/93fe14d/ : FAILURE in 31m 47s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/bc3/870050/1/check/osci-lint/bc391b1/ : FAILURE in 27m 13s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/027/870050/1/check/tox-py38/0278d2b/ : FAILURE in 17m 15s\n- ceph-focal-xena_nova ceph-focal-xena_nova : SKIPPED\n- ceph-focal-yoga_nova ceph-focal-yoga_nova : SKIPPED\n- ceph-jammy-yoga_nova ceph-jammy-yoga_nova : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"07b937668624a9fe4fb223c171d2a61df527d34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2023-01-17 02:40:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e786d0dc1bac3b2c2c1763de95bb1b972dac88ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 02:49:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30ac69fc9bb44526aeed87c3de38a42e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e3de21016f47b7ac251137f88f0aaf : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9d476cc815b43b28f9dac7c8d081a3b : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a95f112d630c46bfaecb8261d5ded156 : FAILURE in 7m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ccf1252c8244d3a80a4903a485f99a8 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6c9afece694b5dd35d560c784dfa108361afddaf","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-01-17 06:17:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/912/870050/2/check/charm-build/9123a66/ : SUCCESS in 8m 28s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/02c/870050/2/check/osci-lint/02c9ff7/ : SUCCESS in 8m 06s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/cd3/870050/2/check/tox-py38/cd3273e/ : SUCCESS in 11m 12s\n- ceph-focal-xena_nova https://openstack-ci-reports.ubuntu.com/artifacts/4fc/870050/2/check/ceph-focal-xena_nova/4fcee53/ : SUCCESS in 1h 32m 12s\n- ceph-focal-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/2fe/870050/2/check/ceph-focal-yoga_nova/2fef6b5/ : SUCCESS in 1h 32m 14s\n- ceph-jammy-yoga_nova https://openstack-ci-reports.ubuntu.com/artifacts/c74/870050/2/check/ceph-jammy-yoga_nova/c74fb3d/ : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a195a95b189035799acc8c8e72465ab088a02e59","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-01-17 08:27:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"54162c529e4be56269f58491733df1dc426975cb","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2023-01-18 02:58:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c7fc127fa790bf43bb235e5ca6bc1b1ee3cb5b5","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-01-18 02:58:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55387ff6901fef319af5c6d9e980661d4d1722bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 02:58:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b12bc8ddcb7736721ae830b4993e9b21ff6af68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 03:08:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd9f95c15384a829b015b1286a69487\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81a34226a184234acc4bb10fe6b935b : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/493ad060cfcb46779bc9a8e1f8ec8fc8 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"eefa7dd535d5ede05e20ef2c5e1901bbfabf959e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 03:08:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bbcc2ec14dd49098786230de0f6acf274a9104d9","revisions":{"ff09a76f79037ded04abe15cc1ceafbd0c8ff085":{"kind":"REWORK","_number":1,"created":"2023-01-13 02:31:34.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/50/870050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/50/870050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/1"}}},"commit":{"parents":[{"commit":"93b9faff4065ca5c474187c49d70baba446a5eb2","subject":"Move default_ephemeral_format to [DEFAULT] section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/93b9faff4065ca5c474187c49d70baba446a5eb2"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2022-12-01 04:08:15.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2023-01-13 02:31:34.000000000","tz":0},"subject":"Use enabled option for vnc","message":"Use enabled option for vnc\n\nvnc_enabled, novnc_enabled are depreacted but should\nbe in DEFAULT section. Commit [1] moved the deprecated\noptions under vnc group so they dont have any effect.\nThis leads to vnc as True which is default option\nin nova.\n\nChange the template to use enabled option instead of\ndeprecated ones\n\n[1] https://opendev.org/openstack/charm-nova-compute/commit/73edc4f817d0a8c281284a0df850d93ea5427014\n\nCloses-Bug: #1998300\nChange-Id: I4193c042f6b3a55dfb1dc57f0f4d3bc71e19006f\n(cherry picked from commit 83497e833d2d76687868552ef84db1906472fd6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/ff09a76f79037ded04abe15cc1ceafbd0c8ff085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/ff09a76f79037ded04abe15cc1ceafbd0c8ff085"}]},"branch":"refs/heads/stable/yoga"},"bbcc2ec14dd49098786230de0f6acf274a9104d9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-17 02:40:22.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/50/870050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/50/870050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/50/870050/2"}}},"commit":{"parents":[{"commit":"7f5e718dcd7200aed499c4eaba9db03fa579ebde","subject":"Pin tox to \u003c 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/7f5e718dcd7200aed499c4eaba9db03fa579ebde"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2022-12-01 04:08:15.000000000","tz":330},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2023-01-17 02:40:22.000000000","tz":0},"subject":"Use enabled option for vnc","message":"Use enabled option for vnc\n\nvnc_enabled, novnc_enabled are depreacted but should\nbe in DEFAULT section. Commit [1] moved the deprecated\noptions under vnc group so they dont have any effect.\nThis leads to vnc as True which is default option\nin nova.\n\nChange the template to use enabled option instead of\ndeprecated ones\n\n[1] https://opendev.org/openstack/charm-nova-compute/commit/73edc4f817d0a8c281284a0df850d93ea5427014\n\nCloses-Bug: #1998300\nChange-Id: I4193c042f6b3a55dfb1dc57f0f4d3bc71e19006f\n(cherry picked from commit 83497e833d2d76687868552ef84db1906472fd6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/bbcc2ec14dd49098786230de0f6acf274a9104d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/bbcc2ec14dd49098786230de0f6acf274a9104d9"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
