)]}'
{"id":"openstack%2Fcharm-nova-compute~889642","triplet_id":"openstack%2Fcharm-nova-compute~master~I7222661017655fd7225db0c677f1a8f5ebb7984d","project":"openstack/charm-nova-compute","branch":"master","topic":"bug/2028559","attention_set":{},"removed_from_attention_set":{"14567":{"account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"last_update":"2023-08-30 13:56:27.000000000","reason":"Change was submitted"},"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2023-08-30 13:44:29.000000000","reason":"\u003cGERRIT_ACCOUNT_935\u003e replied on the change","reason_account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2023-08-17 00:45:35.000000000","reason":"\u003cGERRIT_ACCOUNT_2424\u003e replied on the change","reason_account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2023-08-30 13:56:27.000000000","reason":"Change was submitted"},"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2023-07-25 17:20:07.000000000","reason":"\u003cGERRIT_ACCOUNT_8108\u003e replied on the change","reason_account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"}}},"hashtags":[],"change_id":"I7222661017655fd7225db0c677f1a8f5ebb7984d","subject":"Fix migration across apps when using VMs created from image","status":"MERGED","created":"2023-07-25 17:04:40.000000000","updated":"2023-10-05 09:22:10.000000000","submitted":"2023-08-30 13:56:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"889642-bug/2028559","meta_rev_id":"9005aee4da4efb236ae74807c7712848f863a58f","_number":889642,"virtual_id_number":889642,"owner":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2023-08-28 18:28:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-30 13:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":2,"date":"2023-08-30 13:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":1,"date":"2023-08-30 13:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}],"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 17:15:27.000000000","updated_by":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"reviewer":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"state":"CC"},{"updated":"2023-07-25 17:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 18:44:30.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"CC"},{"updated":"2023-07-25 19:50:39.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-31 15:30:52.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"},{"updated":"2023-08-17 00:45:35.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REMOVED"},{"updated":"2023-08-17 00:45:35.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"}],"messages":[{"id":"48ab05b03447a961e93f2a22d7866a9a711f4118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-07-25 17:04:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2218989dfa222b3b6e05c0ec3786dc7dbeb509a","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2023-07-25 17:15:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81a00db22d2f7a1017fe799f30000683839ed24b","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-07-25 17:17:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"989d737dd6f23ad855c6908988ef0c91cbf0986a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:17:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45a4a20242e54768971275c6b084983d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab0953e4ca340f7bfd27a81f1831bec : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38d9bef3f1a461ba58680e498d5b637 : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9a270b92b714a28b4a94588ac8397bb : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5141ec25b5eb42f7bd680a3674dc700c : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b70966f3b3b464b5c80352aca20713383838e6c3","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2023-07-25 17:18:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20c972a4c215017a00aa7934343f896bc98516fe","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2023-07-25 17:20:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc3760c78145e15938588222b2333ff606850a85","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2023-07-25 18:44:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa59bf18583673e4987153fae8819b368b3a1cbf","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-07-25 19:50:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/772/889642/1/check/charm-build/77229f1/ : SUCCESS in 12m 14s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/8de/889642/1/check/osci-lint/8de5d5c/ : SUCCESS in 8m 29s\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/529/889642/1/check/lunar-antelope/5295846/ : FAILURE in 27m 03s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/62a/889642/1/check/jammy-antelope/62addd6/ : SUCCESS in 1h 22m 51s\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/37f/889642/1/check/jammy-zed/37f4db5/ : FAILURE in 2h 32m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ae00714405bd9f4a8f8d997a376d074d8bba3f0e","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-07-26 13:47:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8db76885564cf9416810c30bbfe08e9558254645","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-07-26 13:49:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59e5089c414b17c1be1158dc9173d4374330b1e7","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-07-26 13:49:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"010f62bacc705042d6f94b2380e2354fec002c5c","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-07-27 02:51:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/d16/889642/1/check/charm-build/d16b42f/ : SUCCESS in 12m 43s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/a44/889642/1/check/osci-lint/a442a59/ : SUCCESS in 8m 48s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/d01/889642/1/check/mantic-bobcat/d012c12/ : FAILURE in 14m 06s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/842/889642/1/check/jammy-bobcat/84247c9/ : FAILURE in 13m 37s (non-voting)\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/f86/889642/1/check/lunar-antelope/f86bda9/ : FAILURE in 25m 40s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/cd9/889642/1/check/jammy-antelope/cd97816/ : SUCCESS in 1h 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"615445ee25d4579eb874febc675068911696e0e5","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-07-31 15:30:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"313adea6cf834d66a5cedeadd90be140be326b28","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-07 15:12:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7f0fa39329c7578d33a71946ba474cbf116c5b4","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-08-07 15:15:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab02543fd4b0e4f9d85e0f72a1ab71d89982c432","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-07 15:20:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b20e8853989ec22916b83874f488bc9b290a20e","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-08-07 16:08:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbc9f2faeb9276fc7f7050baa0b65b0569bbe7a2","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-08-07 16:08:52.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"929d42116b748878ffa5d1aa6a4c4a8a1daa6ec9","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-08 14:29:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c421fd9e927206dd428840668aa99c91b604836b","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-10 12:47:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0881c5fd21c60077cedbf50076a25d64d885c934","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-11 11:58:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f249eaf1d8c5dce06e2e7bded00a520223a611e","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2023-08-11 19:34:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9ad72c009aae360fcf0db4493a5fbedb648822e5","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-14 13:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a383fe2a049a9af698250402201e2f981809b86f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-08-17 00:45:35.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_2424\u003e.","accounts_in_message":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}],"_revision_number":1},{"id":"211e2c04a8318191082dcd94b8d3e8732fb17cc9","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-08-22 09:16:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9f3f6a8292de030d52d1cebb20958078094e3df3","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-22 11:48:30.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bfc9fb34d56308464d1b000dbc30e2ba3fd6bba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-22 13:16:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0fc205c234ce894e9f235924a5e95e2b97eafab7","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-08-22 13:28:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/7ef/889642/2/check/charm-build/7ef951e/ : SUCCESS in 10m 47s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/87f/889642/2/check/osci-lint/87ff7b8/ : FAILURE in 7m 22s\n- mantic-bobcat mantic-bobcat : SKIPPED (non-voting)\n- jammy-bobcat jammy-bobcat : SKIPPED (non-voting)\n- lunar-antelope lunar-antelope : SKIPPED (non-voting)\n- jammy-antelope jammy-antelope : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"6a6551219e96b7b3faf7743370c29353967abf55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:29:23.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e43058a4875146d2aed1e1de8e79f2db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e108f8b7a8004b7f952f21c6a332a751 : FAILURE in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f38fcb372d04d728b72374d4a1e0c4a : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb0207a36b124e52bdfb69ca4a63c26e : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81236dbb86e74009b30c8cfc92459fb4 : FAILURE in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa4f1a63ca494d3f89f3f3e6da653e4c : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"687fc50a9f5650708b7422012ed1b0a96c1666bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-22 13:33:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"59d8336ae1c812a08f98eba17dd33ba26b8c1e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:01:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faad5153cd03487e93e36f255ae7f368\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47c080f5bfa470dbf45454e4ff5a66f : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bab70aa74901450a8a35afd02b725a5d : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f1e347281494e9ead079b8daed8af48 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3c5f329ac8549098d5e8b6caa916a29 : FAILURE in 4m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8d4112c73e949bf9f950fde32f61ffa : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":3},{"id":"dbd85cf7a95a756fa50612cc9fd573061990caf6","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-08-22 18:13:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/2a3/889642/3/check/charm-build/2a3879c/ : SUCCESS in 10m 31s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/34a/889642/3/check/osci-lint/34a8e54/ : SUCCESS in 7m 32s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/d8f/889642/3/check/mantic-bobcat/d8f8ffe/ : FAILURE in 16m 50s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/87d/889642/3/check/jammy-bobcat/87d8a7e/ : FAILURE in 30m 10s (non-voting)\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/d79/889642/3/check/lunar-antelope/d798996/ : TIMED_OUT in 4h 28m 34s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/e5e/889642/3/check/jammy-antelope/e5edddd/ : FAILURE in 1h 18m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8688f36140c5546e624e327e2da70f4db0e374f8","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-23 11:11:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9398a71dcf508e2613cadfde8f35cc54d171057","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-23 11:11:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e9ba1f17ca37d4647eeddecb4476e62e13aeae5","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-08-23 13:29:13.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/02a/889642/3/check/charm-build/02a68bb/ : SUCCESS in 11m 13s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/abf/889642/3/check/osci-lint/abfd906/ : SUCCESS in 8m 06s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/f9b/889642/3/check/mantic-bobcat/f9b1408/ : FAILURE in 15m 18s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/3c5/889642/3/check/jammy-bobcat/3c5161c/ : FAILURE in 29m 47s (non-voting)\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/95e/889642/3/check/lunar-antelope/95eff69/ : FAILURE in 1h 55m 54s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/5c2/889642/3/check/jammy-antelope/5c2c83e/ : FAILURE in 1h 14m 05s","accounts_in_message":[],"_revision_number":3},{"id":"38cc8d0da899c96092484cea025d569490162ab3","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-23 13:58:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0878e8d790245c8d911e7e2e652b94e0e0515032","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-24 12:38:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c143f328dbf9bd0f213ead0ee5463e010cc1b04d","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-28 15:40:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3ac2bcfc4a16ef99552e49b6f5ed428f212c361","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-08-28 15:40:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"517cbb2137d2cbeec805bbfc51bab1e1c4f32377","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 18:28:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/242/889642/3/check/charm-build/2423748/ : SUCCESS in 8m 32s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/771/889642/3/check/osci-lint/7714013/ : SUCCESS in 7m 59s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/435/889642/3/check/mantic-bobcat/435d8b5/ : FAILURE in 17m 44s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/9e6/889642/3/check/jammy-bobcat/9e65bac/ : FAILURE in 36m 41s (non-voting)\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/727/889642/3/check/lunar-antelope/7275c75/ : FAILURE in 2h 39m 19s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/c61/889642/3/check/jammy-antelope/c618e58/ : SUCCESS in 1h 51m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f5ebde6519e70c55a16543884941b3775c742576","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-08-30 13:44:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f9c4b900598942be039404821834d2803960dd8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:44:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8011cd2ad0062f7d96f03353bfc0d7ebb0c49c2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:56:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ebb4652fb245f0b1085032f6eea557\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da8cb70e6f964288817d53b5a64ca953 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dafa7638d1d04503afeb30bacd2ac450 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85e5c9e5ee734c5e9d807b80e8e49b22 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"13f470b08505b88d651b23331c10e7e6bca24a16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:56:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9005aee4da4efb236ae74807c7712848f863a58f","tag":"autogenerated:gerrit:revert","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-10-05 09:22:10.000000000","message":"Created a revert of this change as I21c127aa565e489ba4d93a1efc8ddba63ef32e87","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c3c2cf0349c086dad7f23b180c3ee9ea0f865e8f","revisions":{"b07a27bade007bda7353c07d295fd4fdc8a1aeb9":{"kind":"REWORK","_number":1,"created":"2023-07-25 17:04:40.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/42/889642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/42/889642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/1"}}},"commit":{"parents":[{"commit":"71df10c44b0868e3b952ec534de59c2f9b0378bb","subject":"Merge \"Add docs key and point at Discourse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/71df10c44b0868e3b952ec534de59c2f9b0378bb"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-07-25 17:02:30.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-07-25 17:02:30.000000000","tz":-180},"subject":"[WIP] Fix migration across apps when using ceph ephemeral","message":"[WIP] Fix migration across apps when using ceph ephemeral\n\nChange app name sent to ceph to \u0027nova-compute\u0027,\nallowing all nova-compute apps to use the same\nceph auth.\n\nCloses-bug: #2028559\nChange-Id: I7222661017655fd7225db0c677f1a8f5ebb7984d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/b07a27bade007bda7353c07d295fd4fdc8a1aeb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/b07a27bade007bda7353c07d295fd4fdc8a1aeb9"}]},"branch":"refs/heads/master"},"a9b04cb46494df86f4e2f0faa016091339c22f7c":{"kind":"REWORK","_number":2,"created":"2023-08-22 13:16:57.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/42/889642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/42/889642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/2"}}},"commit":{"parents":[{"commit":"191670b0aab089b7c10a64573d0f83962c38e94d","subject":"Merge \"Add 2023.2 Bobcat support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/191670b0aab089b7c10a64573d0f83962c38e94d"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-07-25 17:02:30.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-08-22 13:16:30.000000000","tz":-180},"subject":"Fix migration across apps when using VMs created from image","message":"Fix migration across apps when using VMs created from image\n\nWhen using the config libvirt-image-backend\u003drbd, VMs\ncreated from image have their disk data stored in ceph\ninstead of the compute node itself.\n\nWhen performing live-migrations, both nodes need to\naccess the same ceph credentials to access the VM\u0027s\ndisk in ceph, but this is currently not possible\nif the nodes involved pertain to different\nnova-compute charm apps.\n\nThis patch changes app name sent to ceph to \u0027nova-compute\u0027,\nallowing all nova-compute apps to use the same ceph auth.\n\nThis patch also includes the charmhelpers sync\nfrom PR: #840\n\nCloses-bug: #2028559\nChange-Id: I7222661017655fd7225db0c677f1a8f5ebb7984d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/a9b04cb46494df86f4e2f0faa016091339c22f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/a9b04cb46494df86f4e2f0faa016091339c22f7c"}]},"branch":"refs/heads/master"},"c3c2cf0349c086dad7f23b180c3ee9ea0f865e8f":{"kind":"REWORK","_number":3,"created":"2023-08-22 13:33:15.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/42/889642/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/42/889642/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/42/889642/3"}}},"commit":{"parents":[{"commit":"191670b0aab089b7c10a64573d0f83962c38e94d","subject":"Merge \"Add 2023.2 Bobcat support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/191670b0aab089b7c10a64573d0f83962c38e94d"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-07-25 17:02:30.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-08-22 13:33:06.000000000","tz":-180},"subject":"Fix migration across apps when using VMs created from image","message":"Fix migration across apps when using VMs created from image\n\nWhen using the config libvirt-image-backend\u003drbd, VMs\ncreated from image have their disk data stored in ceph\ninstead of the compute node itself.\n\nWhen performing live-migrations, both nodes need to\naccess the same ceph credentials to access the VM\u0027s\ndisk in ceph, but this is currently not possible\nif the nodes involved pertain to different\nnova-compute charm apps.\n\nThis patch changes app name sent to ceph to \u0027nova-compute\u0027,\nallowing all nova-compute apps to use the same ceph auth.\n\nThis patch also includes the charmhelpers sync\nfrom PR: #840\n\nCloses-bug: #2028559\nChange-Id: I7222661017655fd7225db0c677f1a8f5ebb7984d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/c3c2cf0349c086dad7f23b180c3ee9ea0f865e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/c3c2cf0349c086dad7f23b180c3ee9ea0f865e8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
