)]}'
{"id":"openstack%2Fcharm-nova-compute~896155","triplet_id":"openstack%2Fcharm-nova-compute~master~I2d27c70f9452b9426a94164eb6c80352a9502a05","project":"openstack/charm-nova-compute","branch":"master","topic":"bug/2037003","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2023-09-25 12:38:06.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"14567":{"account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"last_update":"2023-10-09 14:37:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2d27c70f9452b9426a94164eb6c80352a9502a05","subject":"[WIP] Retain old ceph credentials","status":"ABANDONED","created":"2023-09-21 18:18:21.000000000","updated":"2023-10-09 14:37:02.000000000","total_comment_count":12,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"502d3411138e04da1ee350973bdb967bd90b0646","_number":896155,"virtual_id_number":896155,"owner":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-09-22 22:12:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-22 20:30:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}],"REVIEWER":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 18:29:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 20:03:46.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 23:24:00.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"CC"},{"updated":"2023-09-22 12:19:13.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"CC"}],"messages":[{"id":"b66f27e79842f178d71ea54a5541e60c3f850b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-09-21 18:18:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"573834f62c5c068212c03ff760e20025f5918c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 18:29:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed11b8cc44e48369d652ddaf30464b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e58091da6f4e4c86812ae4c1a3780cb3 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d153ad292e5547829c31e916279a0afb : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25ccffc9eff045faa9bfe7da5e9264ae : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5bf14e9a8ed47e6b195e851edb626ce : FAILURE in 4m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c340dabfb3745a69f593aa879d789b8 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2cf0661dbea89647c908712d7be30fdf2e97c778","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-09-21 20:03:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/4cb/896155/1/check/charm-build/4cbf9ca/ : SUCCESS in 8m 39s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/bae/896155/1/check/osci-lint/bae5535/ : SUCCESS in 5m 33s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/155/896155/1/check/mantic-bobcat/155e8de/ : FAILURE in 16m 06s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/712/896155/1/check/jammy-bobcat/7126935/ : FAILURE in 36m 21s (non-voting)\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/b6f/896155/1/check/lunar-antelope/b6f5c24/ : FAILURE in 52m 29s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/71d/896155/1/check/jammy-antelope/71d6d97/ : SUCCESS in 1h 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6ce5518d4d8f645fd7b9a1bc155e8ca35dfe159d","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2023-09-22 12:19:13.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bfce838f0d20fd1327abb10dd30a4222f134f792","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-09-22 12:24:37.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c232b46d7154cf4564b6dc2d985cd36f4d094e65","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2023-09-22 12:45:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf28f9fde0ec70e4932aa61939a8b878a720e51e","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-09-22 12:54:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9b257dfcfc5e713fe1b68c5c8188007233cebbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-09-22 20:17:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"567d78d57a236604dcbb769fccce3649f4c15f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 20:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/651eb841e0a44cf08e5f6a94ff6d8595\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06249c3731b64ddaad36de47d9195ffb : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5abbc48035554eee9f32fc4dfb2c7556 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c89239944974cab927adca08b7a387b : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/943aa325f7cb43c0b6412279f663a377 : FAILURE in 6m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92a64658993d46bb8abfe54dabd3225f : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"46f5e1dfed66e8ae36c20122a0a6a71d7e466ca2","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-09-22 22:12:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/497/896155/2/check/charm-build/4976795/ : SUCCESS in 6m 44s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/74d/896155/2/check/osci-lint/74d98cf/ : SUCCESS in 5m 44s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/e86/896155/2/check/mantic-bobcat/e8672f7/ : FAILURE in 14m 36s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/d2f/896155/2/check/jammy-bobcat/d2f8c3c/ : FAILURE in 42m 05s (non-voting)\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/d64/896155/2/check/lunar-antelope/d64dfb2/ : FAILURE in 1h 47m 19s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/d99/896155/2/check/jammy-antelope/d991a4b/ : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d01e96cf0cc723be4047a4c78d410934d3a6b561","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2023-09-25 12:38:06.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"502d3411138e04da1ee350973bdb967bd90b0646","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-10-09 14:37:02.000000000","message":"Abandoned\n\nabandoning in favor of https://review.opendev.org/c/openstack/charm-nova-compute/+/897549","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1128e26643e6cfdabf0df3c278f5f514956f4a6f","revisions":{"1d99b70be787438e293c60769b867bb50ac31845":{"kind":"REWORK","_number":1,"created":"2023-09-21 18:18:21.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/55/896155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/55/896155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/1"}}},"commit":{"parents":[{"commit":"ccea3ee4673261368262ac8cf9ee6b429d0db73e","subject":"Merge \"Fix migration across apps when using VMs created from image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/ccea3ee4673261368262ac8cf9ee6b429d0db73e"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-09-21 13:36:05.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-09-21 18:12:03.000000000","tz":-180},"subject":"[WIP] Retain old ceph credentials","message":"[WIP] Retain old ceph credentials\n\nChange [1] migrated the ceph credentials to\n\u0027nova-compute\u0027. Existing nodes would retain\nthe old credentials (in addition to the new one)\nupon upgrading the charm, while newly deployed\nnodes would only have the new credentials. This\nchange ensures newly deployed nodes install\nthe old credentials first on ceph-joined hook,\nand then supercedes it with the new credentials\non ceph-changed hook, therefore also retaining\nthe old credentials.\n\n[1] https://review.opendev.org/889642\n\nPending:\n- Improved unit test coverage\n\nCloses-bug: #2037003\nChange-Id: I2d27c70f9452b9426a94164eb6c80352a9502a05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/1d99b70be787438e293c60769b867bb50ac31845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/1d99b70be787438e293c60769b867bb50ac31845"}]},"branch":"refs/heads/master"},"1128e26643e6cfdabf0df3c278f5f514956f4a6f":{"kind":"REWORK","_number":2,"created":"2023-09-22 20:17:47.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/55/896155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/55/896155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/55/896155/2"}}},"commit":{"parents":[{"commit":"ccea3ee4673261368262ac8cf9ee6b429d0db73e","subject":"Merge \"Fix migration across apps when using VMs created from image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/ccea3ee4673261368262ac8cf9ee6b429d0db73e"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-09-21 13:36:05.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2023-09-22 20:17:26.000000000","tz":-180},"subject":"[WIP] Retain old ceph credentials","message":"[WIP] Retain old ceph credentials\n\nChange [1] migrated the ceph credentials to\n\u0027nova-compute\u0027. Existing nodes would retain\nthe old credentials (in addition to the new one)\nupon upgrading the charm, while newly deployed\nnodes would only have the new credentials. This\nchange ensures newly deployed nodes install\nthe old credentials first on ceph-joined hook,\nand then supercedes it with the new credentials\non ceph-changed hook, therefore also retaining\nthe old credentials.\n\n[1] https://review.opendev.org/889642\n\nPending:\n- Functional tests\n\nCloses-bug: #2037003\nChange-Id: I2d27c70f9452b9426a94164eb6c80352a9502a05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/1128e26643e6cfdabf0df3c278f5f514956f4a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/1128e26643e6cfdabf0df3c278f5f514956f4a6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
