)]}'
{"id":"openstack%2Fcharm-nova-compute~955090","triplet_id":"openstack%2Fcharm-nova-compute~master~I74ecf45c8a7a127de2f903d6783473b45e6ea219","project":"openstack/charm-nova-compute","branch":"master","topic":"bug/2069607","attention_set":{},"removed_from_attention_set":{"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2025-07-16 16:58:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74ecf45c8a7a127de2f903d6783473b45e6ea219","subject":"Force config drive on s390x architecture","status":"MERGED","created":"2025-07-15 17:12:46.000000000","updated":"2025-07-16 16:58:48.000000000","submitted":"2025-07-16 16:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955090-bug/2069607","meta_rev_id":"67498aa68937e270f18bdc1ae7d939f241d1971b","_number":955090,"virtual_id_number":955090,"owner":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-16 16:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":2,"date":"2025-07-16 16:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":1,"date":"2025-07-16 16:53:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-15 17:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 16:53:14.000000000","updated_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"reviewer":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"state":"REVIEWER"}],"messages":[{"id":"b6d8bea40dfa32c977fad70f33f94cf307dd55c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2025-07-15 17:12:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f26a6e7645e3a87461d957a65549809b8d18c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 17:22:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd9d69dc03804c6e8fe1e292449bc39a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038f8981f98546c89a60420b47806714 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5336b06c697b44ae8e1644ef9c29191f : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eabb1e6db77248949229efb82e37136a : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"361a9825f4e55ef6a2d1b8c3c3b396b1d528ddb3","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2025-07-16 16:53:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbf341dea4d6d284f88a3d4373b486959ed8bad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:53:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25e0da192614f45cc24b1c51fb212eb9b85ed960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:58:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d2a847f0c7f4988a2d6a7b27d001ec8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218a9a8d3dde41eea14a99049a493b51 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b473733818c4d31a8672a1ce34fb2ce : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"67498aa68937e270f18bdc1ae7d939f241d1971b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:58:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bc0ef9b57def6158cd771576a5283bd1d62dc7a","revisions":{"7bc0ef9b57def6158cd771576a5283bd1d62dc7a":{"kind":"REWORK","_number":1,"created":"2025-07-15 17:12:46.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/90/955090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-nova-compute","ref":"refs/changes/90/955090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/90/955090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/90/955090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-nova-compute refs/changes/90/955090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-nova-compute refs/changes/90/955090/1"}}},"commit":{"parents":[{"commit":"25a0aba59d0b760b1c12b2c99c269aad80d2e087","subject":"Make cloud-compute hostname same as nova.conf host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/25a0aba59d0b760b1c12b2c99c269aad80d2e087"}]}],"author":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2025-07-14 15:00:20.000000000","tz":-240},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2025-07-15 16:58:27.000000000","tz":-240},"subject":"Force config drive on s390x architecture","message":"Force config drive on s390x architecture\n\nWhen running on s390x force the use of config drive, this is because\non that architecture the DMI data won\u0027t be included in the libvirt\u0027s\ndomain definition, this will prevent cloud-init from knowing the\nplatform where the virtual machine is running on and it will default\nto the config drive mechanism.\n\nChange-Id: I74ecf45c8a7a127de2f903d6783473b45e6ea219\nCloses-Bug: #2069607\nSigned-off-by: Felipe Reyes \u003cfelipe.reyes@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/7bc0ef9b57def6158cd771576a5283bd1d62dc7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-nova-compute/commit/7bc0ef9b57def6158cd771576a5283bd1d62dc7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
