)]}'
{"id":"openstack%2Fcharm-swift-proxy~847513","triplet_id":"openstack%2Fcharm-swift-proxy~master~I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d","project":"openstack/charm-swift-proxy","branch":"master","topic":"batch-update","attention_set":{},"removed_from_attention_set":{"20870":{"account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"last_update":"2022-09-02 00:49:47.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2022-09-21 16:26:55.000000000","reason":"Change was submitted"},"11805":{"account":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"last_update":"2022-09-21 16:26:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d","subject":"Add Kinetic and Zed support","status":"MERGED","created":"2022-06-24 01:27:02.000000000","updated":"2022-09-21 16:26:55.000000000","submitted":"2022-09-21 16:26:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"847513-batch-update","meta_rev_id":"adb9fa7708ce8f930c45cefd24730db79413bd85","_number":847513,"virtual_id_number":847513,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2022-09-06 17:31:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 16:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":2,"date":"2022-09-21 16:17:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":1,"date":"2022-09-21 16:17:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 01:42:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 06:06:24.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 15:48:53.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"},{"updated":"2022-09-21 16:17:04.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"}],"messages":[{"id":"7762c4c6788a0a4daff0d51e6f4adab6f1ef8831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-06-24 01:27:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ad2a4ed2726bdcd0435b09dd14673727a6afed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 01:42:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e4a2ba22dc460c92d4279370e00460 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08cbd2314c0a4ac190561a6b51163ce0 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/215ac16543e54cb69cb9892b9a0c3f9c : FAILURE in 3m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a96bda784b48453d978457bc08b1bb34 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a439e1b1cda791df5cb400bbd0f210ffb60848cc","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-24 06:06:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/12d/847513/1/check/charm-build/12d3fa4/ : SUCCESS in 9m 32s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/c9c/847513/1/check/osci-lint/c9cd4d3/ : SUCCESS in 7m 15s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/234/847513/1/check/tox-py38/23474f0/ : SUCCESS in 10m 21s\n- tox-py310 https://openstack-ci-reports.ubuntu.com/artifacts/5dc/847513/1/check/tox-py310/5dc5b47/ : RETRY_LIMIT in 7m 01s\n- test-s3api-focal-xena test-s3api-focal-xena : SKIPPED\n- test-s3api-focal-yoga test-s3api-focal-yoga : SKIPPED (non-voting)\n- test-s3api-impish-xena test-s3api-impish-xena : SKIPPED (non-voting)\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85eddb450354505d4f39fd9ade7636c4c280bdd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-07-18 13:15:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d4e28be1f1c480a552c2f29aba4bda6417e1e73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 13:31:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e4b7fc454343c89118d572bd1bd081\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/563565a98dc64ccf87c1d8f9108e1617 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e7d920fd9a848bc9c69ab32579923ce : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b19278fb2f2f47fdaa55dd09bb3c7e05 : FAILURE in 3m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/953f261aecdc4351aef4180fb5e94ed8 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"5d0ce1d61c5cd7052cb8d2619b9bc54195d825ae","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-19 07:44:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/cd0/847513/2/check/charm-build/cd07f33/ : SUCCESS in 8m 49s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/af8/847513/2/check/osci-lint/af89323/ : SUCCESS in 6m 52s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/1e6/847513/2/check/tox-py38/1e63ea0/ : SUCCESS in 8m 21s\n- test-s3api-focal-xena https://openstack-ci-reports.ubuntu.com/artifacts/245/847513/2/check/test-s3api-focal-xena/2456804/ : SUCCESS in 39m 08s\n- test-s3api-focal-yoga https://openstack-ci-reports.ubuntu.com/artifacts/514/847513/2/check/test-s3api-focal-yoga/514f2d4/ : FAILURE in 31m 31s (non-voting)\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/98c/847513/2/check/test-s3api-jammy-yoga/98c9652/ : FAILURE in 35m 59s (non-voting)\n- test-s3api-jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/b14/847513/2/check/test-s3api-jammy-zed/b145ea3/ : FAILURE in 24m 26s (non-voting)\n- test-s3api-kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/f85/847513/2/check/test-s3api-kinetic-zed/f85ca94/ : FAILURE in 25m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30d4cb5ebad746169266ed9b46087efc8137005a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-07-22 17:45:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d5de742a07310c29eb650e184b0a162ae6570dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 17:54:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a06cc798b253489790b8b839a60c6d52\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0279d71a501740c59c2c71cdf2e6b5ca : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0d7165bcabe47e8b031c1d5c1d5eed1 : FAILURE in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90b86babf02c45cf9d47a1f2f7095e1d : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"56efc75b7ef1fe9bf0e409a6bdb900d281e64693","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-22 18:44:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/309/847513/3/check/charm-build/3092a5f/ : FAILURE in 5m 31s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/717/847513/3/check/osci-lint/7170a82/ : SUCCESS in 7m 30s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/c0a/847513/3/check/tox-py38/c0a5870/ : SUCCESS in 11m 44s\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb6cfa78e863f2e0c2ad47589c60359fae152d15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-07-25 21:04:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f0204b86c7a4fc9b42e86ac676d235e565d30aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 21:24:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/211adebe28b640b2a407fe5bf569f9fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb666f5ef3f04bb8a05030a14f923550 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fde033f21a034f5a9b792d40c12c2e11 : FAILURE in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15ec8aa8ffcb464686eaf450a04f4ac3 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"1ebef105e2c7c5ad9d540353c6460b23a17e3ee1","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-25 22:34:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/8df/847513/4/check/charm-build/8dfd184/ : FAILURE in 5m 00s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/410/847513/4/check/osci-lint/410937c/ : SUCCESS in 7m 25s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/b04/847513/4/check/tox-py38/b04a9b4/ : SUCCESS in 8m 16s\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5e257aeb85b848b0c02b565f7e306c4f9090fe01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-07-26 12:45:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5443fd2f3457b2e5346f1a23a6fa847f44e94044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 12:55:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99be63b76f4e46b4be1ef440f7208711\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b56d0ab93e6b4653a3256272e45bd06c : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c85fcfa91d55461097044b5d85dd2c42 : SUCCESS in 9m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beec07581d3d45da9ced6a147ed639d4 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":5},{"id":"f6982ecbe95260346ac8812dcb8a6f4ae0cd8d2f","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-26 13:03:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/ce5/847513/5/check/charm-build/ce55f6d/ : FAILURE in 5m 13s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/eb3/847513/5/check/osci-lint/eb30408/ : SUCCESS in 8m 14s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/77d/847513/5/check/tox-py38/77db124/ : SUCCESS in 9m 37s\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a7db76c8f2a8e4d56b568d569d81ca7332e49588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-07-26 19:46:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9d177824e3674d16bec8f3c53a67b78c03d52807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 20:07:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffae2e3b387b407196d3e7a0ab0cee79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14175035ab714c538cbf8047c124daaa : FAILURE in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f98dda5963304b5b93cc7e4a8988e7ca : SUCCESS in 12m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/978770246ade43d7a4fbab3f43c82971 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":6},{"id":"9000aad5dba2514deee25b73ba94234ec818f0c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-07-26 20:32:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"de43f5e1f7c384a99988188c0b475c4664cd3707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 20:50:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d83ecbd900a43ef9b72bd10eba12018\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e48b4c67da14751b74561476a68533d : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0d819bb9e2849fe943ee764884e4919 : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/563e4a6b96b646039f30f0937a6eea65 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":7},{"id":"6f6bf9e4968f0aedb0dc0efbc675c7b1e319b6e4","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-26 22:47:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/454/847513/7/check/charm-build/4540a15/ : FAILURE in 6m 03s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/435/847513/7/check/osci-lint/4358110/ : SUCCESS in 7m 22s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/c57/847513/7/check/tox-py38/c574932/ : FAILURE in 19m 04s\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"efb351f99e3e32b5d768b569b11a602f6591d853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-02 14:59:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"aaaa8375866c933f124056a2092c35ab3d175700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 15:20:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8000615ac3024ba6b039dddc2cd8a793\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d89bbdd47cbd4fe3a65f51ef0345b44e : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51c7deb0cc6847128fa8cd2b481140b3 : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0baf1096efa74526b9b1ce3cdc5b4e2d : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":8},{"id":"133817cc9b6651c5062c4e702fa8543e4078d3e2","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-02 17:24:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/56d/847513/8/check/charm-build/56d4036/ : POST_FAILURE in 9m 07s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/313/847513/8/check/osci-lint/3132f73/ : SUCCESS in 7m 18s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/c6d/847513/8/check/tox-py38/c6d39c1/ : FAILURE in 17m 30s\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"459d07644a8576468c7bf226311ab4b2133ca8ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-04 16:25:35.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent, and commit message as Patch Set 8.","accounts_in_message":[],"_revision_number":9},{"id":"85468dce22787cdb40655c38d974b982684aebc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:51:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a1ffc81f684d368a94685e0094ce4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e764ccb4e18a4ebeacf43d7d9bcae3b0 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f22b17bedc644c598bbe5d42884e90a5 : SUCCESS in 11m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4374fe83e7f64e858b96222c75c68b70 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":9},{"id":"9761d51ed37c2607b532c4080c036acf29c9e29e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-04 18:59:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/e77/847513/9/check/charm-build/e773b72/ : SUCCESS in 22m 05s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/056/847513/9/check/osci-lint/0567213/ : SUCCESS in 11m 03s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/87c/847513/9/check/tox-py38/87c79d7/ : FAILURE in 24m 32s\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"aff53cf1b135e139c319f4f9c5d1a4b5df167092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-05 19:22:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"05c344449a691b73c7309e2c2dbacbeba44e5a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:44:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe4a4e2666f4c23a2dcb1c9c8dec4bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5c0a0a56e84c4abac34a1fd0814ab4 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72b06962ae8c45f296260b12ea981793 : SUCCESS in 9m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6cd1a23e1c74826816e92cf5cf93e05 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":10},{"id":"575df2f6c67a4f44d645f8755c6906cc5f26167c","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-05 22:00:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/7d0/847513/10/check/charm-build/7d0efc6/ : SUCCESS in 8m 06s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/9da/847513/10/check/osci-lint/9dadb1b/ : SUCCESS in 7m 27s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/ede/847513/10/check/tox-py38/edef963/ : FAILURE in 16m 43s\n- test-s3api-jammy-yoga test-s3api-jammy-yoga : SKIPPED (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0a8267af84642886585be7dc6d4416be96138e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-15 20:47:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"549610b8bc0227127f1fac335d284e461f8cbcff","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-15 20:48:01.000000000","message":"Patch Set 11: Verified-1\n\nUnable to freeze job graph: Job test-s3api-jammy-yoga depends on tox-py38 which was not run.","accounts_in_message":[],"_revision_number":11},{"id":"cc0efdf564d0005f006495426ef546214c828506","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-15 20:49:54.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"db3764c7980e6968e0dbd947c8c82054e3de44f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 21:13:18.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3bc4a915a7b4ebe809fde82b1631eff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed1bb6ac6e0f4d00835dc26bae463e52 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/018f96e274d742258e06507ebbd0699f : SUCCESS in 10m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c776a5fbddcf49aaa8505e1b4190adcb : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":12},{"id":"4474092d76cfa9ac4dc6e3a04482edb3568e77f5","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-16 07:39:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/232/847513/12/check/charm-build/23259b4/ : SUCCESS in 7m 54s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/bec/847513/12/check/osci-lint/bec10ae/ : SUCCESS in 7m 39s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/a61/847513/12/check/test-s3api-jammy-yoga/a616ed6/ : FAILURE in 36m 52s (non-voting)\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3e6637cbca3d2d2373b2d42b39fcbb20bc5114bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-16 14:09:17.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f1685ad257b6490b02a52abef83791ba51951c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 14:19:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7caae35dc014449a921a693f1ea49823\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfdcd14b38e4266aee52bacdb562055 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/714bfb8abcd04c539e69645866586a38 : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb467b929f514d2380b10339f028905d : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":13},{"id":"2da069cb23fffdf8634b01231f8bdb230cf0b040","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-16 15:04:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/65f/847513/13/check/charm-build/65f3542/ : SUCCESS in 8m 40s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/d3c/847513/13/check/osci-lint/d3c5378/ : SUCCESS in 7m 28s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/5ac/847513/13/check/test-s3api-jammy-yoga/5ac6b81/ : FAILURE in 36m 17s\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9462a5c76292ce20bd38d35424ade79ab394f59c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-17 18:22:21.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"655f24611f427e39772b59317b1002e8a264a41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 18:51:02.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b7cef1e4cc4e508c2bca23dd594ade\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca26fb22855d4fb59aed0e34b59d58d8 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e25414a3aff4054968999beb2ddf93b : SUCCESS in 10m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bce2a1fc494c4b5a86db125e7a5b5c73 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":14},{"id":"59319c0bf53dd22226ed5d8501de34536ab56738","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-18 06:25:20.000000000","message":"Patch Set 14: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/106/847513/14/check/charm-build/1066cf1/ : SUCCESS in 8m 34s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/5ff/847513/14/check/osci-lint/5ff9e53/ : SUCCESS in 7m 19s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/d28/847513/14/check/test-s3api-jammy-yoga/d28f68c/ : FAILURE in 39m 32s\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"01ed372b69f01b740775674086379c488d7508d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-22 21:12:29.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"51efe6224d16a2fda2a1c4be5d8c0c93d32da3f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 21:39:30.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8fb3567094f4cf5960f8a13f0a9c319\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1547026e78194d74bb11a159338e5baa : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a7dfac9357b4decb2b17d82c08fa6be : SUCCESS in 7m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/703beb850e00486292113e215b699669 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":15},{"id":"cc6f0c46fc63c1cc4ebbba9ae997c3777002836e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-23 07:13:59.000000000","message":"Patch Set 15: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/584/847513/15/check/charm-build/58427bb/ : SUCCESS in 8m 15s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/7e2/847513/15/check/osci-lint/7e21f8e/ : SUCCESS in 6m 53s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/9f2/847513/15/check/test-s3api-jammy-yoga/9f20974/ : FAILURE in 31m 41s\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4ce2d9846fa4f1fb45efcb5adb133a4ab245a684","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-24 18:15:05.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ebd156dbd757b141a85d7ef5f23dc21b13914fb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-26 18:45:28.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"dcc2aa91952397b1f371aa74398c5d58c0e5a23c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:11:00.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d6aefa247c4bbeb22f97c64fdc05a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc56b07ea71a43c5858de832e45d3d09 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/402244d0dcf84fb4b4693848170d5b83 : SUCCESS in 9m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81e501a8124943be86f036e0340ea29f : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":16},{"id":"3a71edcc0bb9d9a899e83878d76431cf2cb4f60b","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-27 07:12:56.000000000","message":"Patch Set 16: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/124/847513/16/check/charm-build/12495a7/ : SUCCESS in 8m 35s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/0f5/847513/16/check/osci-lint/0f5fc52/ : SUCCESS in 8m 01s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/3f2/847513/16/check/test-s3api-jammy-yoga/3f2241b/ : FAILURE in 34m 25s\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d83c92ef246938abd544c22972d407a0cec195b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-08-31 19:33:39.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c5a6f1fe2472afd1029e756e7e973012c49fbe85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 19:43:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f26fda94b24023a6a131c131ec0580\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98113af5369c464ab4a1104749850767 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b45c1533c1b49faa3469c4e69142096 : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e95d4571777467aa7a9933672ed7e78 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":17},{"id":"c29fa84d82dfec0fe532fd37f1021e64cc7bad54","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-09-01 10:20:14.000000000","message":"Patch Set 17: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/11f/847513/17/check/charm-build/11fe313/ : SUCCESS in 8m 52s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/66e/847513/17/check/osci-lint/66e8f1d/ : SUCCESS in 7m 47s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/234/847513/17/check/test-s3api-jammy-yoga/234618b/ : FAILURE in 38m 39s\n- test-s3api-jammy-zed test-s3api-jammy-zed : SKIPPED (non-voting)\n- test-s3api-kinetic-zed test-s3api-kinetic-zed : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1b3a38e1ae24820fe5a6cdac82f0073a7106d8d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-09-02 00:36:08.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"3a23cd9146ca98e59dddf51d6978b93b83d350c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 00:49:47.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc5956de9694d82a7004afb68407365\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64699814c2bc49ef9166ff4836e33867 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0830b7618c6b442b9b428a849ec60492 : SUCCESS in 8m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/422e57505f264ef3886974c3cdbe9e76 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":18},{"id":"8b0556d767fd411bb874982a1b125e7ede1d2afe","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-09-02 02:26:03.000000000","message":"Patch Set 18: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/b0b/847513/18/check/charm-build/b0b68ad/ : SUCCESS in 8m 22s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/2a5/847513/18/check/osci-lint/2a588d7/ : SUCCESS in 8m 15s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/8f3/847513/18/check/test-s3api-jammy-yoga/8f3858e/ : SUCCESS in 55m 03s\n- test-s3api-jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/de6/847513/18/check/test-s3api-jammy-zed/de6a5be/ : FAILURE in 26m 21s (non-voting)\n- test-s3api-kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/ec1/847513/18/check/test-s3api-kinetic-zed/ec17218/ : FAILURE in 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"36b147bd83bdc34413633120854a5ba63d9c5ad8","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-09-06 15:48:53.000000000","message":"Patch Set 18: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"74cc2963a11b31642199307d7968154a58ae947b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-09-06 15:52:08.000000000","message":"Topic bug/1988491 removed","accounts_in_message":[],"_revision_number":18},{"id":"0934dfec4f05b1e208837d92019deda351074732","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-09-06 15:52:13.000000000","message":"Topic set to batch-update","accounts_in_message":[],"_revision_number":18},{"id":"a984f1ddc3eaa7646a0b79dcf9ef0cc975e0d09b","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-09-06 16:19:09.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"75fda033f988e68ea4421f013f2dac37f0b8f826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-09-06 16:23:23.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"277a4a851267f7b5060b4a2431c984b3fadd689b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:51:35.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff1b635ceff4fa3b5c2058c521a4f80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d19a86f23e47fb85a016d823dc2135 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91876fdf55014e93812807cd15ec79aa : SUCCESS in 9m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6d7205082b5425487f99e2171d39e6c : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":19},{"id":"81e183dcffd227cb576539aa5daeba5ffb2ff279","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-09-06 17:31:01.000000000","message":"Patch Set 19: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/a27/847513/19/check/charm-build/a271c77/ : SUCCESS in 8m 50s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/aad/847513/19/check/osci-lint/aad74ca/ : SUCCESS in 6m 39s\n- test-s3api-jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/73d/847513/19/check/test-s3api-jammy-yoga/73d5998/ : SUCCESS in 30m 48s\n- test-s3api-jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/408/847513/19/check/test-s3api-jammy-zed/408ed04/ : FAILURE in 25m 34s (non-voting)\n- test-s3api-kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/b0f/847513/19/check/test-s3api-kinetic-zed/b0ffd2d/ : FAILURE in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b36b77e1d441ecf78170d5c3c51959eca0354a70","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-09-21 15:52:52.000000000","message":"Topic bug/1988491 removed","accounts_in_message":[],"_revision_number":19},{"id":"c88f53deb143afd3c368b2956681c38c9c900ca6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-09-21 15:52:56.000000000","message":"Topic set to batch-update","accounts_in_message":[],"_revision_number":19},{"id":"5d1f6d112eb4664c35f699604e3053237ac8ab2d","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-09-21 16:17:04.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3ebf26434c48b2f4b949db9cf2fb47719991492a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 16:17:31.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"43ab85eea9d5fb83e0c9c08cfca9e0dab237e9f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 16:26:55.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d60cc12fd7c431c835b009fe9c80154\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/341c32632a49404698bc750f17e399e0 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da21823920c549f8a2700102d4aaf8d0 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":19},{"id":"adb9fa7708ce8f930c45cefd24730db79413bd85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 16:26:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"745a89a7c1b1837b72bf36bf9ff2024d40de8fac","revisions":{"915ed768e7162a545f3760ea7a936a93e19d304c":{"kind":"REWORK","_number":1,"created":"2022-06-24 01:27:02.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/1"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-24 01:19:40.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* rebuild for reactive charms\n* change openstack-origin/source default to zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* add run-on base for 22.10\n* drop series from metadata.yaml in favor of run-on bases\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/915ed768e7162a545f3760ea7a936a93e19d304c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/915ed768e7162a545f3760ea7a936a93e19d304c"}]},"branch":"refs/heads/master"},"6116723539479b0b21eff674d25c07041ba4c9d3":{"kind":"REWORK","_number":2,"created":"2022-07-18 13:15:24.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/2"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-07-13 16:43:50.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* rebuild for reactive charms\n* change openstack-origin/source default to zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* add run-on base for 22.10\n* drop series from metadata.yaml in favor of run-on bases\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6116723539479b0b21eff674d25c07041ba4c9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6116723539479b0b21eff674d25c07041ba4c9d3"}]},"branch":"refs/heads/master"},"6854b60422809622c8584ed2a7edc61dff653017":{"kind":"REWORK","_number":3,"created":"2022-07-22 17:45:28.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/3"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-07-22 17:45:11.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* rebuild for reactive charms\n* change openstack-origin/source default to zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* add run-on base for 22.10\n* drop series from metadata.yaml in favor of run-on bases\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6854b60422809622c8584ed2a7edc61dff653017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6854b60422809622c8584ed2a7edc61dff653017"}]},"branch":"refs/heads/master"},"f939e687618695e36d57f1c2f2d48b188a342338":{"kind":"REWORK","_number":4,"created":"2022-07-25 21:04:35.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/4"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-07-25 20:41:06.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/f939e687618695e36d57f1c2f2d48b188a342338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/f939e687618695e36d57f1c2f2d48b188a342338"}]},"branch":"refs/heads/master"},"eb980bf88676aab7d7646ec919e8b525e18826e4":{"kind":"REWORK","_number":5,"created":"2022-07-26 12:45:40.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/5"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-07-26 12:45:20.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/eb980bf88676aab7d7646ec919e8b525e18826e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/eb980bf88676aab7d7646ec919e8b525e18826e4"}]},"branch":"refs/heads/master"},"25fc10b5229711e17d17e7df7525441af1b3a5ad":{"kind":"REWORK","_number":6,"created":"2022-07-26 19:46:52.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/6"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-07-26 19:38:55.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/25fc10b5229711e17d17e7df7525441af1b3a5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/25fc10b5229711e17d17e7df7525441af1b3a5ad"}]},"branch":"refs/heads/master"},"ad05f93cb7ae8e2fa1b69779b2c847d519856cf0":{"kind":"REWORK","_number":7,"created":"2022-07-26 20:32:50.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/7"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-07-26 20:29:00.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/ad05f93cb7ae8e2fa1b69779b2c847d519856cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/ad05f93cb7ae8e2fa1b69779b2c847d519856cf0"}]},"branch":"refs/heads/master"},"6c3f6cef26e8717746dc0b8226cd235991e587db":{"kind":"REWORK","_number":8,"created":"2022-08-02 14:59:32.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/8"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-02 14:50:22.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 1.7/stable\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6c3f6cef26e8717746dc0b8226cd235991e587db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6c3f6cef26e8717746dc0b8226cd235991e587db"}]},"branch":"refs/heads/master"},"123de9dfd4e3e61b8cb67378728c18f7703c343b":{"kind":"NO_CHANGE","_number":9,"created":"2022-08-04 16:25:35.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/9"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-04 16:02:18.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 1.7/stable\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/123de9dfd4e3e61b8cb67378728c18f7703c343b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/123de9dfd4e3e61b8cb67378728c18f7703c343b"}]},"branch":"refs/heads/master"},"b0916ae8bff6d4dac649b3cbef9122fca43041a7":{"kind":"REWORK","_number":10,"created":"2022-08-05 19:22:30.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/10"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-05 18:17:33.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/b0916ae8bff6d4dac649b3cbef9122fca43041a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/b0916ae8bff6d4dac649b3cbef9122fca43041a7"}]},"branch":"refs/heads/master"},"58b64cd78c20b1971c4e2f3a5fc288055bde4eda":{"kind":"REWORK","_number":11,"created":"2022-08-15 20:47:32.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/11"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-15 20:18:51.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/58b64cd78c20b1971c4e2f3a5fc288055bde4eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/58b64cd78c20b1971c4e2f3a5fc288055bde4eda"}]},"branch":"refs/heads/master"},"92dc273df5774044f98b0ca0947def72abc9b876":{"kind":"REWORK","_number":12,"created":"2022-08-15 20:49:54.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/12"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-15 20:49:50.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/92dc273df5774044f98b0ca0947def72abc9b876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/92dc273df5774044f98b0ca0947def72abc9b876"}]},"branch":"refs/heads/master"},"c12165a3395eacc9c57374237a4736a789292601":{"kind":"REWORK","_number":13,"created":"2022-08-16 14:09:17.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/13"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-16 14:09:11.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/c12165a3395eacc9c57374237a4736a789292601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/c12165a3395eacc9c57374237a4736a789292601"}]},"branch":"refs/heads/master"},"6139fe11388a9240e24a8c3465f763625fd68828":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2022-08-17 18:22:21.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/14"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-17 18:05:56.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n* move interface/layer env vars to charmcraft.yaml\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6139fe11388a9240e24a8c3465f763625fd68828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/6139fe11388a9240e24a8c3465f763625fd68828"}]},"branch":"refs/heads/master"},"551edc5609414d431897f237f374720b21a702a5":{"kind":"REWORK","_number":15,"created":"2022-08-22 21:12:29.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/15"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-22 21:07:27.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n* move interface/layer env vars to charmcraft.yaml\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/551edc5609414d431897f237f374720b21a702a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/551edc5609414d431897f237f374720b21a702a5"}]},"branch":"refs/heads/master"},"5052833c17e8ddaa7a05776789a991e1e757ccc5":{"kind":"REWORK","_number":16,"created":"2022-08-26 18:45:28.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/16"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-08-26 18:40:44.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n* move interface/layer env vars to charmcraft.yaml\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/5052833c17e8ddaa7a05776789a991e1e757ccc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/5052833c17e8ddaa7a05776789a991e1e757ccc5"}]},"branch":"refs/heads/master"},"b88b9d363a70a7aec5e13e16cea77a1a128d6d2e":{"kind":"REWORK","_number":17,"created":"2022-08-31 19:33:39.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/13/847513/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/17"}}},"commit":{"parents":[{"commit":"9d55ad9570e8888454b0f1807cde3136f324695f","subject":"Trivial grammar fix in README.md","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/9d55ad9570e8888454b0f1807cde3136f324695f"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2022-08-31 19:33:32.000000000","tz":60},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n* move interface/layer env vars to charmcraft.yaml\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/b88b9d363a70a7aec5e13e16cea77a1a128d6d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/b88b9d363a70a7aec5e13e16cea77a1a128d6d2e"}]},"branch":"refs/heads/master"},"1e63d553aa9234cb92d0b69d18ff2dcd71dba7c9":{"kind":"REWORK","_number":18,"created":"2022-09-02 00:36:08.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/18"}}},"commit":{"parents":[{"commit":"dde918015e2c1c3bfab0f68246b0bb439dd22a97","subject":"Merge \"Add new version template: yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/dde918015e2c1c3bfab0f68246b0bb439dd22a97"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-09-02 00:35:34.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n* move interface/layer env vars to charmcraft.yaml\n* add domain/project to auth section of yoga proxy-server.conf\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\nCloses-Bug: #1988491\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/1e63d553aa9234cb92d0b69d18ff2dcd71dba7c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/1e63d553aa9234cb92d0b69d18ff2dcd71dba7c9"}]},"branch":"refs/heads/master"},"745a89a7c1b1837b72bf36bf9ff2024d40de8fac":{"kind":"REWORK","_number":19,"created":"2022-09-06 16:23:23.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/847513/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-swift-proxy","ref":"refs/changes/13/847513/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-swift-proxy refs/changes/13/847513/19"}}},"commit":{"parents":[{"commit":"fa61a4e59bf35d276b07c243a811b4f934fc244b","subject":"Pin tempest for different python versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/fa61a4e59bf35d276b07c243a811b4f934fc244b"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-06-10 20:14:49.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2022-09-06 16:23:07.000000000","tz":0},"subject":"Add Kinetic and Zed support","message":"Add Kinetic and Zed support\n\n* sync charm-helpers to classic charms\n* change openstack-origin/source default to zed\n* align testing with zed\n* add new zed bundles\n* add zed bundles to tests.yaml\n* add zed tests to osci.yaml and .zuul.yaml\n* update build-on and run-on bases\n* add bindep.txt for py310\n* sync tox.ini and requirements.txt for ruamel\n* use charmcraft_channel 2.0/stable\n* drop reactive plugin overrides\n* move interface/layer env vars to charmcraft.yaml\n* add domain/project to auth section of yoga proxy-server.conf\n\nChange-Id: I93da9ce52bed2b3e8a467491d2a61bfd2ed4ea7d\nCloses-Bug: #1988491\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/745a89a7c1b1837b72bf36bf9ff2024d40de8fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-swift-proxy/commit/745a89a7c1b1837b72bf36bf9ff2024d40de8fac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}}]}],"submit_requirements":[]}
