)]}'
{"id":"openstack%2Fcharm-tempest~746103","triplet_id":"openstack%2Fcharm-tempest~stable%2F20.08~I5d156a567d2f91e03ac7cc504b2d254e7d0820dd","project":"openstack/charm-tempest","branch":"stable/20.08","topic":"release-20.08","hashtags":[],"change_id":"I5d156a567d2f91e03ac7cc504b2d254e7d0820dd","subject":"Updates for stable branch creation","status":"MERGED","created":"2020-08-13 10:40:51.000000000","updated":"2020-08-13 12:08:34.000000000","submitted":"2020-08-13 12:08:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746103-1597320514057-1213e955","meta_rev_id":"59e1bcb679836b5fedbde56b1eda5f5150d88b1a","_number":746103,"virtual_id_number":746103,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 12:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 11:59:14.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 11:59:14.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 11:59:14.000000000","updated_by":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"reviewer":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"state":"REVIEWER"},{"updated":"2020-08-13 12:08:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8847afe30c77eef9386e58425071f796d6116152","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-08-13 10:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5353ebecd33e1daeaff33e39ff74dc600265007f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 10:50:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f11856090bd64ec3b6ca23176bdce55a : SUCCESS in 3m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6c20efa7e2244b8c8921a00cf21a5211 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/757e5707df1e49c7b56f90a2d6144f2a : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f10a7d019aee4410bebe4887dbfae1f7 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1178bfa527c0435eb88998218efc0843 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4286faacf866e5f6c1a48f69824aef45d2e9e960","author":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"date":"2020-08-13 11:59:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"bcc694a00c82ef71888865d351fdff17f9766e38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 11:59:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be72f3813060311a2e67eaaf17f115992708b8c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:08:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83db32c64804932913b87bc61687f0f : SUCCESS in 3m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a60d16ec5d994a84a893cbf305d1d83c : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0562486673b4842b96da5f231b00718 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0906837feb5c46c5b09187183a0ef8f0 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e4c16978a9b92e599e1ea75d24a056747b1c6e04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:08:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61de0e377e924bb80ff98f56b2efbf6d1cfd23bc","revisions":{"61de0e377e924bb80ff98f56b2efbf6d1cfd23bc":{"kind":"REWORK","_number":1,"created":"2020-08-13 10:40:51.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/03/746103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-tempest","ref":"refs/changes/03/746103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-tempest refs/changes/03/746103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-tempest refs/changes/03/746103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-tempest refs/changes/03/746103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-tempest refs/changes/03/746103/1"}}},"commit":{"parents":[{"commit":"64cc0c998aa17dad6e81334a9cdf3be4a6c1d948","subject":"Release sync for 20.08","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-tempest/commit/64cc0c998aa17dad6e81334a9cdf3be4a6c1d948"}]}],"author":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:40:40.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:40:49.000000000","tz":60},"subject":"Updates for stable branch creation","message":"Updates for stable branch creation\n\nSet default branch for git review/gerrit.\n\nSwitch tests to stable.\n\nSwitch to using stable charm-helpers branch.\n\nSwitch to using stable charm.openstack branch.\n\nChange-Id: I5d156a567d2f91e03ac7cc504b2d254e7d0820dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-tempest/commit/61de0e377e924bb80ff98f56b2efbf6d1cfd23bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-tempest/commit/61de0e377e924bb80ff98f56b2efbf6d1cfd23bc"}]},"branch":"refs/heads/stable/20.08"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
