)]}'
{"id":"openstack%2Fcinder-specs~651480","triplet_id":"openstack%2Fcinder-specs~master~Ib5050f5b23300414a7903b48917460639e12aab7","project":"openstack/cinder-specs","branch":"master","topic":"bp/untyped-volumes-default-volume-type","hashtags":[],"change_id":"Ib5050f5b23300414a7903b48917460639e12aab7","subject":"Untyped vol to default vol type","status":"MERGED","created":"2019-04-10 09:32:19.000000000","updated":"2019-07-25 20:33:16.000000000","submitted":"2019-07-25 20:33:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":53,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651480-1564086796824-02757940","meta_rev_id":"45a13bc090d6b0fa087b9ff3975d2c45ddbf9a6c","_number":651480,"virtual_id_number":651480,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":2,"date":"2019-07-25 20:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"date":"2019-07-13 07:34:07.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-25 20:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-07-16 16:58:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2019-07-17 15:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-25 20:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-10 13:24:16.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2019-04-12 10:21:58.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2019-04-16 17:28:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-16 23:21:59.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-04-17 16:24:11.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-03 16:48:54.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2019-06-27 18:49:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-07-16 16:58:32.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2019-07-17 15:24:43.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-07-25 20:22:51.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-07-25 20:33:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13a57502b811bcd237987eda818f762c61fdf044","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-10 09:32:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58b0d84ac043b23e3cd97a31ece6028e1dc77b8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 09:42:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/1/check/openstack-tox-docs/196ee22/html/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/1/check/openstack-tox-pep8/b0c82a9/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"28ab6840fb6b677e8c48673a7ff995d4c8c009b8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-04-10 13:33:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"799436155e4a11686cbaff62235a6541a0bff304","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-04-10 13:33:40.000000000","message":"Patch Set 1:\n\nThanks for getting this spec going!","accounts_in_message":[],"_revision_number":1},{"id":"9fd8adbcf27f82376b4093e80e7fe2d4e96a7ae3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-10 14:42:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"867d5689023e702db08d03682c822afae6493744","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-10 14:48:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"134180db5f7068bb0b9fafb8c2a125a14f27d80c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-10 14:50:25.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nHi Eric,\n\nThanks for the review.","accounts_in_message":[],"_revision_number":3},{"id":"f0f552e8da1e968c5be326d938f9f09a0d22d920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 14:59:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/3/check/openstack-tox-docs/301fc62/html/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/3/check/openstack-tox-pep8/83128bb/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"8e690f5221c888fdd410c0bd5680e49c8bb908a0","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-04-12 10:21:58.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0a8c9926d13eb95c663aea77683355f9de2d52ce","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-15 10:11:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7356ec67d558e81791608ff2f3dafb66f1cb09b8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-15 10:15:32.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nThanks for the review Gorka.","accounts_in_message":[],"_revision_number":4},{"id":"4f7c96ef0a8e5589f43ff71d602d11b89d6d42ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 10:22:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/4/check/openstack-tox-docs/e4f04a0/html/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/4/check/openstack-tox-pep8/8696ed4/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":4},{"id":"c371841643007662c0df8364eb2fe5ddcc12db48","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-04-16 23:21:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d52ef9a46dab0db7379313f15b13e0b7299b72d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-17 07:22:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f70250bb03307b14f49525216654c540a34a73ee","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-17 17:10:03.000000000","message":"Patch Set 4:\n\nSo the meeting discussion [1] led to the conclusion that the type name is not of the the highest priority for now and reviews can be focused more on generalizing the whole idea behind the implementation.\n\n[1] http://eavesdrop.openstack.org/meetings/cinder/2019/cinder.2019-04-17-16.00.log.html#l-228","accounts_in_message":[],"_revision_number":4},{"id":"a52ced9da806cb0300d26dab24e7ea1b3ea268ad","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-24 16:00:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"736b88437ee001f1236b5007e166c5513dd928d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 16:07:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/5/check/openstack-tox-docs/b574e33/html/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/5/check/openstack-tox-pep8/d7c7cc8/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":5},{"id":"fecd2e481c2fabb04cf27b5bea018c37f981534b","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2019-05-03 16:48:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e23409d6d62ee0c035b63e8ea3a547605dc1b922","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2019-05-03 16:48:54.000000000","message":"Patch Set 5: Code-Review-1","accounts_in_message":[],"_revision_number":5},{"id":"1cb294e06134ec7f390dada737329f5fc3631ac6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-05-20 06:33:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"274f87a8b63779c1709b2c6af5b85b3c0ce26590","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-05-20 17:30:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ca9210001a5a813ae9029af716b2fb914919aea","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-05-20 17:31:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"410a7ea26dac9e85a2d17d31543fbda0c4abad70","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-06-03 05:32:59.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"cba8f57fa785de387981609388a7005d5be51d0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 05:38:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/6/check/openstack-tox-docs/bd5e5d9/html/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/6/check/openstack-tox-pep8/fa165b6/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":6},{"id":"5f91a625c5da22ee017d7699c10d91dd6e628a12","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-06-03 05:38:52.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"12b4c8e6e11b6fe20cf7fee3a1def26f8d06e9ba","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-06-09 21:07:59.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nA few typos that should be addressed.  All-in-all I think this represents what we have talked about.","accounts_in_message":[],"_revision_number":6},{"id":"a8f8a7bb955669579ce9fb8ef967e1ee4385a844","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-06-10 14:15:02.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fcc0383ae2b74147038da609ec75e1d47fe86075","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-06-11 06:25:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3333dfa5a179874ab0e7a95e03c4a0d26804c922","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-06-11 06:28:35.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"81fa8b451dd2d238e1e98a3656b4ce02deba5966","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 06:37:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/7/check/openstack-tox-docs/564163e/html/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/7/check/openstack-tox-pep8/74a25e6/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":7},{"id":"c4b9e07a700d4e9c67d7f539bd5c596acbceaa58","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-06-27 16:50:53.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"a9975dd0c85942ba683ca00c67e7038a07dc5fb4","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-06-27 16:51:36.000000000","message":"Patch Set 8: Code-Review+2\n\nI corrected a couple of typos.  This looks good otherwise.  It is consistent with what I remember us discussing previously.","accounts_in_message":[],"_revision_number":8},{"id":"be13680fd819a3d03271dcd1b95cebc085752d26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 17:00:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/8/check/openstack-tox-docs/486dfa7/html/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/8/check/openstack-tox-pep8/4a72c83/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":8},{"id":"537f1a06d0e8a9fa12dfb6ad0a150445140cab36","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-06-27 18:49:26.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cc7811ea00fde1f493565f4a5bb5a326f9845a76","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-06-28 05:52:49.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2e0ff9c44d8de7e0e6a072cc9400a98ec5855039","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-08 09:01:23.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"0f09d5ef8ee2451620b746ff4743d77276db64ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 09:12:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/9/check/openstack-tox-docs/b451eed/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/9/check/openstack-tox-pep8/7ed7344/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":9},{"id":"f4982a0d067224b55f16c273f1a8185473d2725c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-08 14:52:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e0c05459476dc46245ec6ff7e4ada31d38a73a90","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-09 10:05:25.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5d432b1839a64ad42cf7a2f7f05c9ef4c12eef7e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-10 17:36:12.000000000","message":"Patch Set 9: Code-Review-1\n\n(7 comments)\n\nI\u0027m a bit confused about __DEFAULT__ volume type vs. the value of the default_volume_type config option.  Some questions inline.","accounts_in_message":[],"_revision_number":9},{"id":"407176f8eba6108e3f379fec11acc4211eb7ac9c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-12 06:15:48.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3ade9c685171ebeadc271ff1bbf14f514332fcac","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-12 06:15:58.000000000","message":"Patch Set 9:\n\n(7 comments)\n\nThanks for the review Brian.","accounts_in_message":[],"_revision_number":9},{"id":"71934440533122df548d0e66259631780679b94d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-12 06:21:05.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b153ece747ac2406db0b3eadbe9cd05ead2dda52","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-12 06:21:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7ee065fc6614d43508a71a97a7fa59ce5426948a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 06:29:15.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/11/check/openstack-tox-docs/5566289/ : POST_FAILURE in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/11/check/openstack-tox-pep8/ccfeb55/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":11},{"id":"909f105ebdc75d0726f35bee095da2535eae08c0","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-13 07:34:07.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0e58eddc83801c805b75911d43523e8abe16d5bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 07:42:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/11/check/openstack-tox-docs/c37de08/html/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/11/check/openstack-tox-pep8/92958c7/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":11},{"id":"303cc5287596d342e5798248d83f10b0680c63df","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-16 00:45:10.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nQuestion for Eric inline.","accounts_in_message":[],"_revision_number":11},{"id":"c03b8524e4e1fc96740e06dc1590d979f8a06ad5","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2019-07-16 16:58:32.000000000","message":"Patch Set 11: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"cdd9284be890895a29e9a32aeb81afb80c4ec863","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-17 15:24:43.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)\n\nI am ok with going forward with this.  Would rather not bikeshed on it forever.","accounts_in_message":[],"_revision_number":11},{"id":"818914c6a6ca17ecff2da1020c751ed10899e6c9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-25 20:22:51.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nI think we\u0027re all in favor of doing this, so let\u0027s move it along and figure out whether __DEFAULT__ volume type should be mutable during the implementation phase.  (See my question on PS9, Line 45)","accounts_in_message":[],"_revision_number":11},{"id":"23d5df0073745a910e9f06af54e2df4f419bf78b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 20:23:02.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"45f8f5015f04b50ccf76c824d0073ed1f0d51363","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 20:33:16.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/651480/11/gate/openstack-tox-docs/7321090/html/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/80/651480/11/gate/openstack-tox-pep8/d464b21/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":11},{"id":"2657affaab5dacdc3c4bad26f0854962516fb533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 20:33:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"99edb4e74cc3b1805dc234264387d43a6795302f","revisions":{"0ce421288fe086aba9172da4fc14c55a1a7c7317":{"kind":"REWORK","_number":1,"created":"2019-04-10 09:32:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/1"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:27:30.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0ce421288fe086aba9172da4fc14c55a1a7c7317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0ce421288fe086aba9172da4fc14c55a1a7c7317"}]},"branch":"refs/heads/master"},"fd9e59598f4c150f8a659d12041110a9750903c9":{"kind":"REWORK","_number":2,"created":"2019-04-10 14:42:55.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/2"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 11:03:48.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/fd9e59598f4c150f8a659d12041110a9750903c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/fd9e59598f4c150f8a659d12041110a9750903c9"}]},"branch":"refs/heads/master"},"6a73ad9cf151cde99f5a66cb18138cd78bd3b514":{"kind":"REWORK","_number":3,"created":"2019-04-10 14:48:35.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/3"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 11:09:56.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6a73ad9cf151cde99f5a66cb18138cd78bd3b514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6a73ad9cf151cde99f5a66cb18138cd78bd3b514"}]},"branch":"refs/heads/master"},"306a52d5ea96290d77858780198d53bdaffdcf3b":{"kind":"REWORK","_number":4,"created":"2019-04-15 10:11:24.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/4"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-15 10:10:56.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/306a52d5ea96290d77858780198d53bdaffdcf3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/306a52d5ea96290d77858780198d53bdaffdcf3b"}]},"branch":"refs/heads/master"},"2351643fd1567f699bebd999dcf371f3f719878f":{"kind":"REWORK","_number":5,"created":"2019-04-24 16:00:16.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/5"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-24 15:59:57.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2351643fd1567f699bebd999dcf371f3f719878f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2351643fd1567f699bebd999dcf371f3f719878f"}]},"branch":"refs/heads/master"},"9af8e565e5f3164121c9bb5e244ea1daf304475e":{"kind":"REWORK","_number":6,"created":"2019-06-03 05:32:59.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/6"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2019-06-03 05:32:53.000000000","tz":0},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9af8e565e5f3164121c9bb5e244ea1daf304475e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9af8e565e5f3164121c9bb5e244ea1daf304475e"}]},"branch":"refs/heads/master"},"4039e936c310bc29f47515828807a8e96a2f8f42":{"kind":"REWORK","_number":7,"created":"2019-06-11 06:25:52.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/7"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-06-11 06:25:48.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4039e936c310bc29f47515828807a8e96a2f8f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4039e936c310bc29f47515828807a8e96a2f8f42"}]},"branch":"refs/heads/master"},"ed4ce1d6d8eb8233abc0538067c0adc57773ad72":{"kind":"REWORK","_number":8,"created":"2019-06-27 16:50:53.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/80/651480/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/8"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","date":"2019-06-27 16:50:47.000000000","tz":0},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ed4ce1d6d8eb8233abc0538067c0adc57773ad72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ed4ce1d6d8eb8233abc0538067c0adc57773ad72"}]},"branch":"refs/heads/master"},"0a6641374ecf65d390f4004d468929301fb5f178":{"kind":"REWORK","_number":9,"created":"2019-07-08 09:01:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/9"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2019-07-08 09:01:16.000000000","tz":0},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0a6641374ecf65d390f4004d468929301fb5f178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0a6641374ecf65d390f4004d468929301fb5f178"}]},"branch":"refs/heads/master"},"b113b9abe26d4a0841f3a067dabf0e12654d11cc":{"kind":"REWORK","_number":10,"created":"2019-07-12 06:15:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/10"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-07-12 06:15:24.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b113b9abe26d4a0841f3a067dabf0e12654d11cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b113b9abe26d4a0841f3a067dabf0e12654d11cc"}]},"branch":"refs/heads/master"},"99edb4e74cc3b1805dc234264387d43a6795302f":{"kind":"REWORK","_number":11,"created":"2019-07-12 06:21:05.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/80/651480/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/80/651480/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/80/651480/11"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-04-10 05:24:30.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2019-07-12 06:20:48.000000000","tz":330},"subject":"Untyped vol to default vol type","message":"Untyped vol to default vol type\n\nThis spec proposes to use default volume type instead of\nallowing users to create untyped volumes.\n\nChange-Id: Ib5050f5b23300414a7903b48917460639e12aab7\nPartial-Implements: blueprint untyped-volumes-default-volume-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/99edb4e74cc3b1805dc234264387d43a6795302f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/99edb4e74cc3b1805dc234264387d43a6795302f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
