)]}'
{"id":"openstack%2Fcinder-specs~664973","triplet_id":"openstack%2Fcinder-specs~master~I51929a022764c85b9464205f3fe07ca07803234c","project":"openstack/cinder-specs","branch":"master","hashtags":[],"change_id":"I51929a022764c85b9464205f3fe07ca07803234c","subject":"Volume Rekey spec","status":"MERGED","created":"2019-06-12 16:39:30.000000000","updated":"2019-07-25 20:11:48.000000000","submitted":"2019-07-25 20:11:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664973-1564085508283-62a04a63","meta_rev_id":"971d6e5a0998f1ba6c9a787aa6a3b902a087fc49","_number":664973,"virtual_id_number":664973,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":2,"date":"2019-07-25 20:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-25 20:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2019-07-24 20:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-25 20:02:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 14:43:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-07-13 16:22:13.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-13 18:48:50.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-14 02:36:14.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-14 02:48:10.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-24 20:03:08.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-07-25 20:02:55.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-07-25 20:11:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6f435c7e16a73709f8704a87f098a9de44529f6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-06-12 16:39:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea228191c5a04491a0aa26c977ce0d0e99653802","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-06-12 16:44:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aefacd58d88cc551728ee356827f32573ba2cdc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 16:50:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/664973/1/check/openstack-tox-docs/cc3ac50/html/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/73/664973/1/check/openstack-tox-pep8/c0d3279/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4e8c561a0d5595354e7a9164d6d19187729289fb","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-03 16:20:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9906e57d3232998084250777dfd10d7621a3cfc5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-03 16:47:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcf15fd564c7f631fa2ab3fed12f895159b6257c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:56:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/664973/2/check/openstack-tox-docs/5f301bf/html/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/73/664973/2/check/openstack-tox-pep8/16ffc05/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e7c7810e091fce0f4b9f05fae337a65026f80fd0","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-03 17:05:12.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\nA few in-line questions, but all-in-all I don\u0027t have concerns.","accounts_in_message":[],"_revision_number":2},{"id":"ced04801972671aa2b18d6f0fb4768e15362a424","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-09 20:14:35.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"53d6de7ada5ffc05473ac407a79c8c51a1acf4c8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-10 14:43:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nI have a long inline comment about Jay\u0027s suggestion of an option for this feature, but I think I talked myself out of it, so I\u0027m +2 on this.","accounts_in_message":[],"_revision_number":2},{"id":"b7e82cc4e0fb2d7553bc75a7a9cb5988526e0016","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-10 21:09:21.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nComment about the Barbican \u0027Consumers API\u0027 inline.","accounts_in_message":[],"_revision_number":2},{"id":"53de4d8c92336e9a646f85d1bfb80f81e4dccb37","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 16:22:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"574050acc5c3ae9d412c993e59b3d6a1a9c0a1ee","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 18:48:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"33dbb410721aced7a4c42b92539b4e9dae1c8889","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-14 02:36:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d9944579513467687810acf837d0137d19a4c2ed","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 02:48:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"691af037283fc5ebbce0d1e5c98ce93ecef2ed9c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-18 15:16:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a649c4e7ae82604f902c44a89abee483c78e1b91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 15:27:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/664973/3/check/openstack-tox-docs/0cbf0f7/html/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/73/664973/3/check/openstack-tox-pep8/03e7f0e/ : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"425e43acfed3374157fc62872e78fb2b2cefe722","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-24 16:07:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cc6c3d615fce95e7c710592aeb26d8c36e058406","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 16:16:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/664973/4/check/openstack-tox-docs/b13b620/html/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 http://logs.openstack.org/73/664973/4/check/openstack-tox-pep8/28f91b3/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":4},{"id":"e5eb8d0363957200ad9c97b8de0e9b7623c8990b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-24 20:03:08.000000000","message":"Patch Set 4: Code-Review+2\n\nI am good with this now that it is passing the check.","accounts_in_message":[],"_revision_number":4},{"id":"1f767edd53cff9e8646947f1815a408d978c70e7","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-07-25 20:02:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0417b2311b5e0c9f196ef8666c4238b0501379d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 20:03:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3633f4a673952f08dd38343c33cc3f8a8cfcee70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 20:11:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/664973/4/gate/openstack-tox-docs/a813082/html/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/73/664973/4/gate/openstack-tox-pep8/1204473/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"971d6e5a0998f1ba6c9a787aa6a3b902a087fc49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 20:11:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"569be4724d234318513df529b287c0c65f1a58a7","revisions":{"212e0c907960eac8c50873ca14d7edf541fbb90a":{"kind":"REWORK","_number":1,"created":"2019-06-12 16:39:30.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/73/664973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/73/664973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/1"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-06-12 16:36:46.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-06-12 16:36:46.000000000","tz":-240},"subject":"Volume Rekey spec","message":"Volume Rekey spec\n\nChange encryption keys on volumes.\n\nChange-Id: I51929a022764c85b9464205f3fe07ca07803234c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/212e0c907960eac8c50873ca14d7edf541fbb90a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/212e0c907960eac8c50873ca14d7edf541fbb90a"}]},"branch":"refs/heads/master"},"7a1e9f33dc23a76c1b9f9bdadad3438f0fbeef8a":{"kind":"REWORK","_number":2,"created":"2019-07-03 16:47:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/73/664973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/73/664973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/2"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-06-12 16:36:46.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-07-03 16:47:42.000000000","tz":-240},"subject":"Volume Rekey spec","message":"Volume Rekey spec\n\nChange encryption keys on volumes.\n\nChange-Id: I51929a022764c85b9464205f3fe07ca07803234c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7a1e9f33dc23a76c1b9f9bdadad3438f0fbeef8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7a1e9f33dc23a76c1b9f9bdadad3438f0fbeef8a"}]},"branch":"refs/heads/master"},"521cdf682d3b85ba7282a61022cbbae548a65a86":{"kind":"REWORK","_number":3,"created":"2019-07-18 15:16:00.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/73/664973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/73/664973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/3"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-06-12 16:36:46.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-07-18 15:15:47.000000000","tz":-240},"subject":"Volume Rekey spec","message":"Volume Rekey spec\n\nChange encryption keys on volumes.\n\nChange-Id: I51929a022764c85b9464205f3fe07ca07803234c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/521cdf682d3b85ba7282a61022cbbae548a65a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/521cdf682d3b85ba7282a61022cbbae548a65a86"}]},"branch":"refs/heads/master"},"569be4724d234318513df529b287c0c65f1a58a7":{"kind":"REWORK","_number":4,"created":"2019-07-24 16:07:44.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/73/664973/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/73/664973/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/73/664973/4"}}},"commit":{"parents":[{"commit":"e1aa893acfae5017f0e43611987d1a3f7948cc55","subject":"Merge \"Move Generic Backups spec to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e1aa893acfae5017f0e43611987d1a3f7948cc55"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-06-12 16:36:46.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-07-24 16:07:20.000000000","tz":-240},"subject":"Volume Rekey spec","message":"Volume Rekey spec\n\nChange encryption keys on volumes.\n\nChange-Id: I51929a022764c85b9464205f3fe07ca07803234c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/569be4724d234318513df529b287c0c65f1a58a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/569be4724d234318513df529b287c0c65f1a58a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
