)]}'
{"id":"openstack%2Fcinder-specs~674533","triplet_id":"openstack%2Fcinder-specs~master~Id34aaf2719d157c1f9d1b477840e51b11ff9f02e","project":"openstack/cinder-specs","branch":"master","topic":"bp/sandstone-volume-driver","hashtags":[],"change_id":"Id34aaf2719d157c1f9d1b477840e51b11ff9f02e","subject":"add sandstone volume driver bp.","status":"ABANDONED","created":"2019-08-05 06:13:26.000000000","updated":"2019-12-13 22:17:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"173e018d25c19145118401959b2fdc756572fce8","_number":674533,"virtual_id_number":674533,"owner":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-05 06:19:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-08-05 09:54:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 06:19:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 09:54:57.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"ed1a0c8a7c7a8496265fcd48962636518e28de0a","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2019-08-05 06:13:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7849c5395b3b7d897ba5a45431f3977e2b78998d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 06:19:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/33/674533/1/check/openstack-tox-docs/495db90/html/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://logs.opendev.org/33/674533/1/check/openstack-tox-pep8/e7f9b43/ : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0ceaabad65186378747b9f71f879d0c4ac17de9e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-08-05 09:54:57.000000000","message":"Patch Set 1: Code-Review-2\n\nWe\u0027ve passed milestone-2 and new spec proposal won\u0027t be accepted.\nOnce Train gets released, spec proposal for \u0027U\u0027 release will open and it can be proposed in the respective directory.\nAlso drivers doesn\u0027t require spec, this info can be mentioned in the driver documentation proposed along with driver.","accounts_in_message":[],"_revision_number":1},{"id":"41428ff5dd9db336f494311bdb85192090606866","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 22:17:49.000000000","message":"Abandoned\n\nSpecs are not needed to add a new driver.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd24f461b84ba6f757cd5cf591c42cd4ec1b7df9","revisions":{"cd24f461b84ba6f757cd5cf591c42cd4ec1b7df9":{"kind":"REWORK","_number":1,"created":"2019-08-05 06:13:26.000000000","uploader":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"ref":"refs/changes/33/674533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/33/674533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/33/674533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/33/674533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/33/674533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/33/674533/1"}}},"commit":{"parents":[{"commit":"7f65f08aabeadc229c3896f06444ecb25d457e30","subject":"Merge \"Untyped vol to default vol type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7f65f08aabeadc229c3896f06444ecb25d457e30"}]}],"author":{"name":"yangziguan","email":"yangziguan@szsandstone.com","date":"2019-08-05 06:11:53.000000000","tz":480},"committer":{"name":"yangziguan","email":"yangziguan@szsandstone.com","date":"2019-08-05 06:11:53.000000000","tz":480},"subject":"add sandstone volume driver bp.","message":"add sandstone volume driver bp.\n\nChange-Id: Id34aaf2719d157c1f9d1b477840e51b11ff9f02e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/cd24f461b84ba6f757cd5cf591c42cd4ec1b7df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/cd24f461b84ba6f757cd5cf591c42cd4ec1b7df9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
