)]}'
{"id":"openstack%2Fcinder-specs~674545","triplet_id":"openstack%2Fcinder-specs~master~I7cc6710930bbdb5b697ea261c5960a16b503449f","project":"openstack/cinder-specs","branch":"master","topic":"bp/sandstone-volume-driver","hashtags":[],"change_id":"I7cc6710930bbdb5b697ea261c5960a16b503449f","subject":"elimiate synax error.","status":"ABANDONED","created":"2019-08-05 07:48:36.000000000","updated":"2019-08-22 22:45:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1c28f9afd02baf7e270be0a26b188d21b0c30f3","_number":674545,"virtual_id_number":674545,"owner":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-05 07:54:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-08-05 09:55:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 07:54:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 09:55:35.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"6d811ee6ef9b7e640b79651dd414da2a58186831","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2019-08-05 07:48:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2374408e299c81931031e66f383693ddbe768696","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 07:54:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/45/674545/1/check/openstack-tox-docs/ad53ea5/html/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://logs.opendev.org/45/674545/1/check/openstack-tox-pep8/c751954/ : FAILURE in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"30b4432f1542381f254c892c0c96f0bc5f9b52d2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-08-05 09:55:35.000000000","message":"Patch Set 1: Code-Review-2\n\nShould be updated in the main patch","accounts_in_message":[],"_revision_number":1},{"id":"15859469ecdc0dfb65ab19064a81d838f7c0e0dd","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-08-22 22:45:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d73618b3edfec59f1644f2da3bbae53f6b81125","revisions":{"2d73618b3edfec59f1644f2da3bbae53f6b81125":{"kind":"REWORK","_number":1,"created":"2019-08-05 07:48:36.000000000","uploader":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"ref":"refs/changes/45/674545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/45/674545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/45/674545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/45/674545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/45/674545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/45/674545/1"}}},"commit":{"parents":[{"commit":"cd24f461b84ba6f757cd5cf591c42cd4ec1b7df9","subject":"add sandstone volume driver bp.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/cd24f461b84ba6f757cd5cf591c42cd4ec1b7df9"}]}],"author":{"name":"yangziguan","email":"yangziguan@szsandstone.com","date":"2019-08-05 07:44:32.000000000","tz":480},"committer":{"name":"yangziguan","email":"yangziguan@szsandstone.com","date":"2019-08-05 07:44:32.000000000","tz":480},"subject":"elimiate synax error.","message":"elimiate synax error.\n\nChange-Id: I7cc6710930bbdb5b697ea261c5960a16b503449f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2d73618b3edfec59f1644f2da3bbae53f6b81125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2d73618b3edfec59f1644f2da3bbae53f6b81125"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
