)]}'
{"id":"openstack%2Fcinder-specs~674547","triplet_id":"openstack%2Fcinder-specs~master~Ifad4e1c6117e7c204de2dfbb7bf76cdef378cb68","project":"openstack/cinder-specs","branch":"master","topic":"bp/sandstone-volume-driver","hashtags":[],"change_id":"Ifad4e1c6117e7c204de2dfbb7bf76cdef378cb68","subject":"elimitnate trailing whitespace.","status":"ABANDONED","created":"2019-08-05 08:02:32.000000000","updated":"2019-08-22 22:44:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e16df2d5fe1b03e9a9a056f120872bba675bd96c","_number":674547,"virtual_id_number":674547,"owner":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-05 13:58:59.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-05 08:09:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-08-05 09:55:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 08:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 09:55:23.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2019-08-05 13:58:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"dedd0fe0b4dcb22c1bb33dfd4fa4a1cf4c7ea2ec","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2019-08-05 08:02:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4faf4260baee29745444f9531497ef56f76f31c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 08:09:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/47/674547/1/check/openstack-tox-docs/ed936e1/html/ : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://logs.opendev.org/47/674547/1/check/openstack-tox-pep8/f18a1db/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"177020d43bb1fed0146b3a650903017cda945742","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-08-05 09:55:23.000000000","message":"Patch Set 1: Code-Review-2\n\nShould be updated in the main patch","accounts_in_message":[],"_revision_number":1},{"id":"32eea87dc8d92ad8d1d0b7cb5327813e6b2b0737","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-05 13:58:59.000000000","message":"Patch Set 1:\n\nShould also not be submitted by a CI account.","accounts_in_message":[],"_revision_number":1},{"id":"ebb46fb073cd96c72c33d248031f56fcafa10786","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-08-22 22:44:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51a35dfd785b0322ac47a5e0412181733b23baf2","revisions":{"51a35dfd785b0322ac47a5e0412181733b23baf2":{"kind":"REWORK","_number":1,"created":"2019-08-05 08:02:32.000000000","uploader":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"ref":"refs/changes/47/674547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/47/674547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/47/674547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/47/674547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/47/674547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/47/674547/1"}}},"commit":{"parents":[{"commit":"2d73618b3edfec59f1644f2da3bbae53f6b81125","subject":"elimiate synax error.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2d73618b3edfec59f1644f2da3bbae53f6b81125"}]}],"author":{"name":"yangziguan","email":"yangziguan@szsandstone.com","date":"2019-08-05 08:00:44.000000000","tz":480},"committer":{"name":"yangziguan","email":"yangziguan@szsandstone.com","date":"2019-08-05 08:00:44.000000000","tz":480},"subject":"elimitnate trailing whitespace.","message":"elimitnate trailing whitespace.\n\nChange-Id: Ifad4e1c6117e7c204de2dfbb7bf76cdef378cb68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/51a35dfd785b0322ac47a5e0412181733b23baf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/51a35dfd785b0322ac47a5e0412181733b23baf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
