)]}'
{"id":"openstack%2Fcinder-specs~678250","triplet_id":"openstack%2Fcinder-specs~master~I3f909742501ab3a3f36ba9eee4ede0b814f76f2d","project":"openstack/cinder-specs","branch":"master","topic":"add-untargeted","hashtags":[],"change_id":"I3f909742501ab3a3f36ba9eee4ede0b814f76f2d","subject":"Add untargeted folder for specs","status":"MERGED","created":"2019-08-23 15:04:48.000000000","updated":"2019-08-26 16:57:10.000000000","submitted":"2019-08-26 16:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678250-1566838630853-abdccd44","meta_rev_id":"92363b5d9036270124b0b7b24efc936e90f904ce","_number":678250,"virtual_id_number":678250,"owner":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":2,"date":"2019-08-26 16:57:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28940,"name":"Woojay Poynter","email":"tenaciouswp@gmail.com","username":"woojay"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-26 16:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-26 16:44:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28940,"name":"Woojay Poynter","email":"tenaciouswp@gmail.com","username":"woojay"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-26 16:46:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28940,"name":"Woojay Poynter","email":"tenaciouswp@gmail.com","username":"woojay"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28940,"name":"Woojay Poynter","email":"tenaciouswp@gmail.com","username":"woojay"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 15:48:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-26 15:48:24.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-08-26 16:44:47.000000000","updated_by":{"_account_id":28940,"name":"Woojay Poynter","email":"tenaciouswp@gmail.com","username":"woojay"},"reviewer":{"_account_id":28940,"name":"Woojay Poynter","email":"tenaciouswp@gmail.com","username":"woojay"},"state":"REVIEWER"},{"updated":"2019-08-26 16:46:29.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-08-26 16:57:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25a3202fb47baed18aa15b69091b807d2234917c","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-08-23 15:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e39989a503b17a95e69bf740e3e123d5ce1ea94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:16:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d128d1f7505e4c698ce40d2d04979e01 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615e95602ac447d290a583d2686fac94 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c26695561c3f2b984e0b021ca146a21fba21ab79","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-08-23 17:09:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6e7f64cfb0dd392fcb0889d809919d76e656dca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:17:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/547712d2dba343fda388f6beec960b8c : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0648ef3a0e4db089eae2da21c760bf : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4087a95963b2bc7b51ff8b1007c27a7a8203fc7e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-26 15:48:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c2f428a75c1d60e20e4851b4a2fb7f6e567588e3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-26 16:15:11.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nTypo noted inline","accounts_in_message":[],"_revision_number":2},{"id":"552f340e2703a46a02bb60bc406282ab187b46b1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-08-26 16:16:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f09b73064f9875f6d53bb0f10f5d5cf2b8a2bc0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:32:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e31e7cf97d4250af5bc0c6ecdbcb7f : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e177a696a2a4af5bc766520dd08b906 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":3},{"id":"fe918df5ba254a74d40eee9bff2c97dd573ff0a6","author":{"_account_id":28940,"name":"Woojay Poynter","email":"tenaciouswp@gmail.com","username":"woojay"},"date":"2019-08-26 16:44:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1eb5c669e7aa746597d4b1e484150c0dfe15b6bb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-26 16:46:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nCarrying over Sean\u0027s +2 from PS2 (only change was a typo fix).","accounts_in_message":[],"_revision_number":3},{"id":"64b386f36aba05157d4222a1251f715203cc2c43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:46:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e35a0a44354bc2a51e8081ab800f261b70c4ba75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:57:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b500bc067ce4e6abcf91898e5c6dbcd : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aab9351f4ae48cc8b6df8fa7b78cf87 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e942fe5cc754c5a2b5a9f29055ca4b8e9cd3758d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:57:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6a4aa8c6b352cfcbae198b79dcc3a9ed1a4e35cc","revisions":{"bdc78fd8803409486560f50cec4b45fb64205b9d":{"kind":"REWORK","_number":1,"created":"2019-08-23 15:04:48.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/50/678250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/50/678250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/1"}}},"commit":{"parents":[{"commit":"7f65f08aabeadc229c3896f06444ecb25d457e30","subject":"Merge \"Untyped vol to default vol type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7f65f08aabeadc229c3896f06444ecb25d457e30"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-08-23 15:02:25.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-08-23 15:02:25.000000000","tz":-300},"subject":"Add untargeted folder for specs","message":"Add untargeted folder for specs\n\nIt was discussed at the Cinder Mid-Cycle that we\nneeded something to do with approved specs that were\nstill relevant but not being worked.  The recommendation\nwas to add an untargeted folder like the one that Glance\nuses.  This patch does this.\n\nI also discovered that the last two sections in\nthe index page were not displaying due to\nincorrect RST format.  This patch also fixes\nthat.\n\nChange-Id: I3f909742501ab3a3f36ba9eee4ede0b814f76f2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bdc78fd8803409486560f50cec4b45fb64205b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bdc78fd8803409486560f50cec4b45fb64205b9d"}]},"branch":"refs/heads/master"},"85e7b43286f46059593a594c9484acab9ddb455c":{"kind":"REWORK","_number":2,"created":"2019-08-23 17:09:03.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/50/678250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/50/678250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/2"}}},"commit":{"parents":[{"commit":"7f65f08aabeadc229c3896f06444ecb25d457e30","subject":"Merge \"Untyped vol to default vol type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7f65f08aabeadc229c3896f06444ecb25d457e30"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-08-23 15:02:25.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-08-23 17:08:49.000000000","tz":-300},"subject":"Add untargeted folder for specs","message":"Add untargeted folder for specs\n\nIt was discussed at the Cinder Mid-Cycle that we\nneeded something to do with approved specs that were\nstill relevant but not being worked.  The recommendation\nwas to add an untargeted folder like the one that Glance\nuses.  This patch does this.\n\nI also discovered that the last two sections in\nthe index page were not displaying due to\nincorrect RST format.  This patch also fixes\nthat.\n\nChange-Id: I3f909742501ab3a3f36ba9eee4ede0b814f76f2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/85e7b43286f46059593a594c9484acab9ddb455c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/85e7b43286f46059593a594c9484acab9ddb455c"}]},"branch":"refs/heads/master"},"6a4aa8c6b352cfcbae198b79dcc3a9ed1a4e35cc":{"kind":"REWORK","_number":3,"created":"2019-08-26 16:16:46.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/50/678250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/50/678250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/50/678250/3"}}},"commit":{"parents":[{"commit":"7f65f08aabeadc229c3896f06444ecb25d457e30","subject":"Merge \"Untyped vol to default vol type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7f65f08aabeadc229c3896f06444ecb25d457e30"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-08-23 15:02:25.000000000","tz":-300},"committer":{"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","date":"2019-08-26 16:16:38.000000000","tz":0},"subject":"Add untargeted folder for specs","message":"Add untargeted folder for specs\n\nIt was discussed at the Cinder Mid-Cycle that we\nneeded something to do with approved specs that were\nstill relevant but not being worked.  The recommendation\nwas to add an untargeted folder like the one that Glance\nuses.  This patch does this.\n\nI also discovered that the last two sections in\nthe index page were not displaying due to\nincorrect RST format.  This patch also fixes\nthat.\n\nChange-Id: I3f909742501ab3a3f36ba9eee4ede0b814f76f2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6a4aa8c6b352cfcbae198b79dcc3a9ed1a4e35cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6a4aa8c6b352cfcbae198b79dcc3a9ed1a4e35cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
