)]}'
{"id":"openstack%2Fcinder-specs~679133","triplet_id":"openstack%2Fcinder-specs~master~Ic13e52cdb6d8dd7c83ead34fc2f80b7b03065f5e","project":"openstack/cinder-specs","branch":"master","topic":"move-generic-backups","hashtags":[],"change_id":"Ic13e52cdb6d8dd7c83ead34fc2f80b7b03065f5e","subject":"Move generic backup spec to untargeted","status":"MERGED","created":"2019-08-28 20:04:21.000000000","updated":"2019-09-10 14:46:34.000000000","submitted":"2019-09-10 14:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679133-1568126794842-a9d30349","meta_rev_id":"4c3467364111a1f3f41cc7723cbc593c285f6dd5","_number":679133,"virtual_id_number":679133,"owner":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-09-10 14:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2019-09-10 14:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-08 03:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2019-09-10 14:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-08 03:41:35.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-09-10 14:38:38.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-09-10 14:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d5a6de33ef1e4e40b0e38f285fc91c9b3a114ad","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-08-28 20:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb77208fb24ed5d474cc1f8bc4d1639192fa9639","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3341e92d4435404f9e4606b5502cea4c : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0ccb6cc6c54222a8468dba27de162e : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a4607f5594cd8424e3d53517312be7c43c56ced8","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-09-08 03:41:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7512560dd55f1f9ddd1e5ba8f1fee38f1cb4bd43","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-10 14:38:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nUnfortunately, I proposed required patches too late, so I didn\u0027t get enough feedback to get it implemented:(","accounts_in_message":[],"_revision_number":1},{"id":"b0c228431ece43c31f3bc3d5152397cc12776955","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:38:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e50742994907723a0f0b86d331ac3a6ec960735d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:46:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bf152e6a185357502127e0b86f07648022d08bfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:46:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f2a06e92ac4dfd95bbceb20513d184 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d0cfa1f4c54e46bbbbacd26ec0d8b0 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6df6999f318921d97c3840556590b8445f02ac8","revisions":{"f6df6999f318921d97c3840556590b8445f02ac8":{"kind":"REWORK","_number":1,"created":"2019-08-28 20:04:21.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/33/679133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/33/679133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/33/679133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/33/679133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/33/679133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/33/679133/1"}}},"commit":{"parents":[{"commit":"256a502613f5a5d316f91df243d586b14f3983e9","subject":"Identify Cinder as the Block Storage Service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/256a502613f5a5d316f91df243d586b14f3983e9"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-08-28 20:00:00.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-08-28 20:00:00.000000000","tz":-300},"subject":"Move generic backup spec to untargeted","message":"Move generic backup spec to untargeted\n\nWe have been sliding the generic backup spec along for\na while and I believe that Ivan wasn\u0027t planning to get this\nin for train.  As a result I am going to move this to\nuntargeted and we can put it into a release when we know\nthe code is closer to landing.\n\nChange-Id: Ic13e52cdb6d8dd7c83ead34fc2f80b7b03065f5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f6df6999f318921d97c3840556590b8445f02ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f6df6999f318921d97c3840556590b8445f02ac8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
