)]}'
{"id":"openstack%2Fcinder-specs~689977","triplet_id":"openstack%2Fcinder-specs~master~I962c998ac62f4468a1deebd5757b0e0b628d9a26","project":"openstack/cinder-specs","branch":"master","topic":"bp/volume-record-failure-info","hashtags":[],"change_id":"I962c998ac62f4468a1deebd5757b0e0b628d9a26","subject":"Add features for volume to record failure info","status":"ABANDONED","created":"2019-10-22 06:34:09.000000000","updated":"2020-05-07 09:36:25.000000000","total_comment_count":50,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96abb57cc48760beadae6dff73a952b825f30060","_number":689977,"virtual_id_number":689977,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"date":"2020-02-04 02:43:11.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-20 07:16:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-2,"date":"2020-02-02 20:54:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":-2,"date":"2020-01-28 21:00:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 08:30:56.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-10-22 08:31:13.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2019-10-22 10:03:20.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-10-22 11:24:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-18 20:48:23.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-12-20 07:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-28 21:00:43.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-02-02 20:54:50.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"80dedaab90553bfb85493fea2073ee424610dc57","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-22 06:34:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"616ad9bc5c8287c595ac6c7ebca0149519e8c17c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-22 06:35:01.000000000","message":"Topic set to bp/volume-record-failure-info","accounts_in_message":[],"_revision_number":1},{"id":"5c7b069b15d343c740885b90dd7480807b3066aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 06:44:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850c19391a3c437ebbd042fc63877f5b : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08310ace02ee41b2ac54fceaf955ef0b : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"eb8cb72bf46a4224fa7df71595add92e0cc8b2b9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-22 07:17:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5a7af700c6b5fff5c5f03f5ec8c939cb9f0e746d","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-22 08:22:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06fe4acd851cdc1d7f001fd7a454f9c6d3e82e15","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-22 08:22:49.000000000","message":"Patch Set 2:\n\n(6 comments)\n\nthank for your review,Brin Zhang.","accounts_in_message":[],"_revision_number":2},{"id":"64cb62d17333649658837ca29ba5f34026682a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 08:29:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46dc65f83be43c8aea647bc724f38f7 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/195799d733e847f2b386221f0dcef3c2 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"b5753accdacd76c01b6f7aa533d431d255231333","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-22 10:03:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nhaixin, you should add more details for this SPEC.","accounts_in_message":[],"_revision_number":2},{"id":"a46a4d46dff06b409071dc40681193d38ab3da25","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 11:24:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nHow does this differ from user messages? I think it needs more detail and a better explanation of how this will be used differently than that functionality.","accounts_in_message":[],"_revision_number":2},{"id":"4bf7c3b1f5ee308f1924782da755576064d5d5e7","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-23 09:13:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2372f4d0f92c655820433502f1e23ccc1de76f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:25:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/941a0421bde64a3088ba1d42cdb5f6a6 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/440f42ded7ae473db338ebd6c80fdf72 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"986dbd6157084b36a50466c2f7945a8e1a25631c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-23 10:00:56.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nthank for your review,Sean McGinnis,Brin Zhang.\nI think this feature and user messages have the following differences.\n(1):user messages can not be queried using volume ID, only by message id, If there were a lot of results returned, it would take us a long time to find the error message in the result that we want to see for that volume ID, but this feature will support query by volume ID.\n(2)There may be multiple asynchronous errors in the same volume,user messages will force the user to find the most recent asynchronous error at the time of creation, this feature will return the last asynchronous error message.\n(3)There are 12 actions and 16 details defined in user messages, If a new feature comes along later, We will have to manually extend Their definition, It\u0027s inconvenient. in this feature error messages are automatically captured using function decorators.\n(4)this feature can also captured traceback information(details column in database),This will make it easier for users or developers to locate the error directly.","accounts_in_message":[],"_revision_number":3},{"id":"19c9729b43fa2f12b9dab04e980294e7aa1b68c9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-10-23 17:06:26.000000000","message":"Patch Set 3:\n\n(1):user messages can not be queried using volume ID, only by message id\nThis is not accurate.  User messages can be queried by volume IDs.","accounts_in_message":[],"_revision_number":3},{"id":"24fde92202bb8554ea3ad8f4b3a5afa37e83d304","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-10-23 17:07:50.000000000","message":"Patch Set 3: Code-Review-1\n\nThis looks like it should reuse the user messages framework, with the additional functionality of auto-capturing more failures in the code rather than relying so much on hard-coded messages for particular cases.","accounts_in_message":[],"_revision_number":3},{"id":"28fcafc12cf6b3f60d0dffa200c5d3f847458e17","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-24 02:17:52.000000000","message":"Patch Set 3:\n\nthank for your review,Eric Harney.\nI\u0027ve tried again, user messages can can be queried by volume IDs.\nYes, the framework of this feature is similar to user messages, but it is more extensible than user messages by using decorators to automatically catch asynchronous error messages, rather than relying so much on hard-coded messages for particular cases, I think it\u0027s useful to us.","accounts_in_message":[],"_revision_number":3},{"id":"5aaf78cf9cd8911d4bca95ab0ee25677461938c0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-09 21:58:35.000000000","message":"Patch Set 3: Code-Review-1\n\nI agree with Eric and Sean, it would be better to work this into the user message framework.  Your point that the user messages show you multiple faults, not just the most recent, seems more of a feature than a bug to me.\n\nAdditionally, each of these faults probably *should* be represented by a user message (even if we add a \u0027fault\u0027 element to the get-volume response).\n\nFinally, I think the \u0027detail\u0027 field needs to be observable by an admin only, because it could leak administrative information about the cloud that shouldn\u0027t be exposed to users.","accounts_in_message":[],"_revision_number":3},{"id":"dd4ea6afbc742b3342d32163a199710dcaa9f46e","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-12-18 20:48:23.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)\n\nSeveral issues to address if we go forward with this.  Also sounds like this may not be the right way to implement it.","accounts_in_message":[],"_revision_number":3},{"id":"b9d966c216ee1390e12352e514ece2fd9dd605b8","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-12-20 07:07:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"090b33ced0b85a0cab08db6b97cea4f29513d5b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 07:16:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd47672ce4d4993a2866f65af24a365 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ea2b56a1bc481db4b825ad1e58e35f : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":4},{"id":"80330aec60b3a4150faeae6f1eceae23e7d15819","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-12-20 07:47:35.000000000","message":"Patch Set 4:\n\n(6 comments)\n\nthank for your review, Brian Rosmaita and Jay Bryant.\n   In order to avoid leak administrative information to the end user.\nI\u0027ve optimized this feature, only admin can get attr of fault.\n   This framework is different from user messages, it can auto-capturing\nmore failures in the code rather than relying so much on hard-coded\nmessages for particular cases(as Eric Harney said), Ability to pinpoint\na line in a file.it will help developers or operations staff to locate\ncode errors.","accounts_in_message":[],"_revision_number":4},{"id":"dccabd652a19e13e0e55677e4b6af50cf7527ad2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-28 16:20:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(8 comments)\n\nI\u0027m still not quite sold on this.  A few typos and points I\u0027d like you to address are noted inline.\n\nIt looks like the use case being addressed now is: A customer has a volume in error status and contacts the operator; the operator wants to be able to do a volume-show and see the \u0027fault\u0027 information for the most recent error event that this particular volume had.\n\nI guess this new table won\u0027t get out of hand if we only store the most recent fault for any volume.  But I\u0027m not sure that using something like Logstash might be more useful.","accounts_in_message":[],"_revision_number":4},{"id":"c0e82621a5571ac2dfdd207cab9e86cc11901338","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-28 16:26:07.000000000","message":"Patch Set 4:\n\nI don\u0027t want to give the impression that I\u0027m against this, either.  One good outcome would be that after you\u0027ve implemented this change, it should be easy for someone to follow up and make sure that we\u0027re recording a user message corresponding to each of these faults, which would be nice for end users.","accounts_in_message":[],"_revision_number":4},{"id":"2d390923ec830717fc0030c3caf89f1cc52bfe91","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-01-28 21:00:43.000000000","message":"Patch Set 4: Code-Review-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a9a29959fafbdd6af2c3471276cbe474abb2e5a4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-29 13:15:02.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nEric raises a good point, that it\u0027s not obvious why the current user message framework cannot handle your use case.  See our comments inline, and please address them on your next revision.  Thanks!","accounts_in_message":[],"_revision_number":4},{"id":"9769fad9f4172872c37ac9ebe1612a323127e49b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-02-02 20:54:50.000000000","message":"Patch Set 4: Code-Review-2\n\nThe spec freeze for Ussuri has passed.  If you are still interested in this proposal, please revise and retarget the spec for the Victoria release.","accounts_in_message":[],"_revision_number":4},{"id":"533f5d17f07bdcac02dff2a555e76627721f74c2","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-02-04 02:43:11.000000000","message":"Patch Set 4:\n\n(7 comments)\n\nthank you Eric Harney and Brian Rosmaita.\nIt might be a good idea to optimize the user message framework, so both message and detail are filled in automatically instead of in message_field\nFixed, message populates the captured exception message, and details populates the error stack to help the user quickly retrieve the corresponding error log.","accounts_in_message":[],"_revision_number":4},{"id":"96abb57cc48760beadae6dff73a952b825f30060","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-05-07 09:36:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"956bc80995aff7b54c5af82a4833a9ba61224ef0","revisions":{"0155152b07420a794b4195c08a73c14a09cc142f":{"kind":"REWORK","_number":1,"created":"2019-10-22 06:34:09.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/77/689977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/689977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/1"}}},"commit":{"parents":[{"commit":"165a81dde2c735aa3147d500278f12936e5b7ab9","subject":"Merge \"Added os-brick adjustments in the Image Encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/165a81dde2c735aa3147d500278f12936e5b7ab9"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-22 06:20:50.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-22 06:33:15.000000000","tz":480},"subject":"Add features for volume to record failure info","message":"Add features for volume to record failure info\n\nThis feature will support to record volume error info to the database\nautomatically when volume meet some action error, it will help us to\nfind the failure reason quickly, and save the trouble of finding the\ncorresponding error log in the multi-node cluster.\n\nChange-Id: I962c998ac62f4468a1deebd5757b0e0b628d9a26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0155152b07420a794b4195c08a73c14a09cc142f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0155152b07420a794b4195c08a73c14a09cc142f"}]},"branch":"refs/heads/master"},"8171c816794043af58945979f6b9fc6fe00b627e":{"kind":"REWORK","_number":2,"created":"2019-10-22 08:22:06.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/77/689977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/689977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/2"}}},"commit":{"parents":[{"commit":"165a81dde2c735aa3147d500278f12936e5b7ab9","subject":"Merge \"Added os-brick adjustments in the Image Encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/165a81dde2c735aa3147d500278f12936e5b7ab9"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-22 06:20:50.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-22 08:20:54.000000000","tz":480},"subject":"Add features for volume to record failure info","message":"Add features for volume to record failure info\n\nThis feature will support to record volume error info to the database\nautomatically when volume meet some action error, it will help us to\nfind the failure reason quickly, and save the trouble of finding the\ncorresponding error log in the multi-node cluster.\n\nChange-Id: I962c998ac62f4468a1deebd5757b0e0b628d9a26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8171c816794043af58945979f6b9fc6fe00b627e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8171c816794043af58945979f6b9fc6fe00b627e"}]},"branch":"refs/heads/master"},"ca3e47630c73d1b078448316b5808cdd353001e3":{"kind":"REWORK","_number":3,"created":"2019-10-23 09:13:32.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/77/689977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/689977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/3"}}},"commit":{"parents":[{"commit":"165a81dde2c735aa3147d500278f12936e5b7ab9","subject":"Merge \"Added os-brick adjustments in the Image Encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/165a81dde2c735aa3147d500278f12936e5b7ab9"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-22 06:20:50.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-23 09:10:51.000000000","tz":480},"subject":"Add features for volume to record failure info","message":"Add features for volume to record failure info\n\nThis feature will support to record volume error info to the database\nautomatically when volume meet some action error, it will help us to\nfind the failure reason quickly, and save the trouble of finding the\ncorresponding error log in the multi-node cluster.\n\nChange-Id: I962c998ac62f4468a1deebd5757b0e0b628d9a26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ca3e47630c73d1b078448316b5808cdd353001e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ca3e47630c73d1b078448316b5808cdd353001e3"}]},"branch":"refs/heads/master"},"956bc80995aff7b54c5af82a4833a9ba61224ef0":{"kind":"REWORK","_number":4,"created":"2019-12-20 07:07:28.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/77/689977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/689977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/689977/4"}}},"commit":{"parents":[{"commit":"165a81dde2c735aa3147d500278f12936e5b7ab9","subject":"Merge \"Added os-brick adjustments in the Image Encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/165a81dde2c735aa3147d500278f12936e5b7ab9"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-22 06:20:50.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2019-12-20 07:07:23.000000000","tz":480},"subject":"Add features for volume to record failure info","message":"Add features for volume to record failure info\n\nThis feature will support to record volume error info to the database\nautomatically when volume meet some action error, it will help us to\nfind the failure reasonably quickly, and save the trouble of finding the\ncorresponding error log in the multi-node cluster.\n\nChange-Id: I962c998ac62f4468a1deebd5757b0e0b628d9a26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/956bc80995aff7b54c5af82a4833a9ba61224ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/956bc80995aff7b54c5af82a4833a9ba61224ef0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
