)]}'
{"id":"openstack%2Fcinder-specs~695630","triplet_id":"openstack%2Fcinder-specs~master~Ic7f259c3af6bd691e6d24ed180430e760d28eba8","project":"openstack/cinder-specs","branch":"master","topic":"spec/copy-image-in-multiple-stores","hashtags":[],"change_id":"Ic7f259c3af6bd691e6d24ed180430e760d28eba8","subject":"Copy volume to image in multiple stores of glance","status":"MERGED","created":"2019-11-22 06:44:36.000000000","updated":"2020-01-28 15:19:30.000000000","submitted":"2020-01-28 15:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695630-1580224675579-fa858c1e","meta_rev_id":"914ad8619c5691623d8d73fdba545bba592585c3","_number":695630,"virtual_id_number":695630,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 15:17:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-20 22:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-12-19 21:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-18 20:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2020-01-10 09:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-28 15:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 20:31:31.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-12-20 22:47:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-10 09:29:54.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-01-28 15:10:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-28 15:17:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e6e804a6c1ddd000f865dbfa9052f5725e0912b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-22 06:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"956492c62d3372690b94f42c50b2cf04db026697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 06:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8872017710a4d81be28de0db7a73ce3 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/546da09abe674c8ead82850503d555d2 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"206bc737737a5aceced9660a299c63f7f78d417b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-11-22 10:35:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)\n\nThanks for the spec Abhishek.\nFew NITs and a query inline.","accounts_in_message":[],"_revision_number":1},{"id":"c0981f529443634fbe29bd38faa69193d23899fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-25 05:34:40.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nThank you for review Rajat!!","accounts_in_message":[],"_revision_number":1},{"id":"02eee439af9c0d7fcba126193c986ad2eac088c8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-25 05:37:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e40d1fa257c014c50bb07a3ade0992906c7b750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 05:42:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be0b49c61644a329865c1c922ba878f : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dec26fcc9744a39a0064a1431d2e5f1 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"c7f1ffd37ee848df56e41115a695e2e8e3b52b7c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-26 07:24:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"03e094e35870a7c122d6f8b1d41848db64460378","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-26 07:29:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e2b50aa4dad5aab26296f75d145b4a3661dabd35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 07:37:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62505a67b914dccb8e6c1471fb99be1 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce06ab95799440f386d2642ecadf5df9 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":4},{"id":"0c9a9c24c017aa34e0eb457bd392962d7a0620e3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-26 07:53:13.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e923a17a3cdf8cb8474c15fe6f11565e41edeed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 07:57:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa39200a1ea4381a5c02373473540f4 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20340a4cc9214e209e7a458fc7ae425f : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":5},{"id":"0687befef65902dd60a1c9a4b8a821f75460c445","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 22:16:59.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks good overall.","accounts_in_message":[],"_revision_number":5},{"id":"dabd7321dd4a52ba9b5587c41597fc55e1cf7ac1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-17 13:36:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(7 comments)\n\nSome clarifications requested inline.  But overall, this looks good.  Thanks, Abhishek!","accounts_in_message":[],"_revision_number":5},{"id":"a3ee6f7b282413a375314fcf0736751199510fb7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-18 06:53:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0784d68b55e152435e8f5106734cc8e18d6fca0d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-18 06:55:08.000000000","message":"Patch Set 5:\n\n(7 comments)\n\nBrian, Thank you for review!","accounts_in_message":[],"_revision_number":5},{"id":"a43e9c39bb6122bc3418398730b9d52d3db14af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 07:10:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df79854662e460eb653eab86ac6e92b : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00faf3dcd16a4216a283c5967409a37f : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":6},{"id":"b709cd3408e846a6e54437d69990e42f85f053f7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-12-18 20:31:31.000000000","message":"Patch Set 6: Code-Review+2\n\nThis spec sounds like what we had previously agreed upon.","accounts_in_message":[],"_revision_number":6},{"id":"6eb94c7f0038d08f067213b1808bf5e15de1cd7f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-19 21:48:59.000000000","message":"Patch Set 6: Code-Review+2\n\nRequested clarifications have been made, and this is in line with what we discussed at the PTG.","accounts_in_message":[],"_revision_number":6},{"id":"53e92834bf48f0182b3d915bb25a39cc59e64611","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-20 22:47:24.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks like all comments have been address, and this looks good to me.","accounts_in_message":[],"_revision_number":6},{"id":"1efdc3d5fca810f90b6be51b4a702544381b3bc4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-12-24 09:58:52.000000000","message":"Patch Set 6: Code-Review+1\n\n(6 comments)\n\nThe overall spec makes sense to me however i\u0027ve some corrections(can be done in a followup) \u0026 comments inline.","accounts_in_message":[],"_revision_number":6},{"id":"a779c2e8b02d8bd52d612a37a72e8f86e111cc0c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-10 09:29:54.000000000","message":"Patch Set 6: Code-Review+2\n\nI feel it\u0027s good to go, my comments can be addressed in a followup.\nwill wait for other cores to take a look.","accounts_in_message":[],"_revision_number":6},{"id":"43f12a5453f68239342b7d663193066da96fc24f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-28 15:10:11.000000000","message":"Patch Set 6: Workflow+1\n\nLooks like we have a consensus.","accounts_in_message":[],"_revision_number":6},{"id":"f2d1c219de39589514461ae828c6f1ac0cf50b74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:10:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"18f6152fc7de50b62fee7152dcdfcce5bbd764ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:17:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e106c12033ef4350ba798757d5859d09 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fe9635f70743ee9db6c1e2e5087bd5 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":6},{"id":"221b2ecbed10cda33942090934205050fd14246e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:17:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"34fce54da87f05bba0774565ed11d788c136b62e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:19:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/90fa60dae7114b78ad03f33f515bd1a1 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f4254c0a77b07d511fdb9aa5576b293fbec63426","revisions":{"824f751467b3e1cb7e296c5925daa85eaf6c3119":{"kind":"REWORK","_number":1,"created":"2019-11-22 06:44:36.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/695630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/695630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/1"}}},"commit":{"parents":[{"commit":"99b817c38124729f3b27a6e02320478a2590376b","subject":"Add ussuri directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/99b817c38124729f3b27a6e02320478a2590376b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-22 06:43:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-22 06:44:35.000000000","tz":0},"subject":"Image snapshot co-location","message":"Image snapshot co-location\n\nChange-Id: Ic7f259c3af6bd691e6d24ed180430e760d28eba8\nImplements: image-snapshot-co-location\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/824f751467b3e1cb7e296c5925daa85eaf6c3119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/824f751467b3e1cb7e296c5925daa85eaf6c3119"}]},"branch":"refs/heads/master"},"c117af9edf1f8434e6b0dc26b29fe4e9387101c5":{"kind":"REWORK","_number":2,"created":"2019-11-25 05:37:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/695630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/695630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/2"}}},"commit":{"parents":[{"commit":"99b817c38124729f3b27a6e02320478a2590376b","subject":"Add ussuri directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/99b817c38124729f3b27a6e02320478a2590376b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-22 06:43:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-25 05:37:27.000000000","tz":0},"subject":"Image snapshot co-location","message":"Image snapshot co-location\n\nChange-Id: Ic7f259c3af6bd691e6d24ed180430e760d28eba8\nImplements: image-snapshot-co-location\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c117af9edf1f8434e6b0dc26b29fe4e9387101c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c117af9edf1f8434e6b0dc26b29fe4e9387101c5"}]},"branch":"refs/heads/master"},"992a21caa28ed0b34e6f675fdaeb3a73b52a8c29":{"kind":"REWORK","_number":3,"created":"2019-11-26 07:24:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/695630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/695630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/3"}}},"commit":{"parents":[{"commit":"99b817c38124729f3b27a6e02320478a2590376b","subject":"Add ussuri directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/99b817c38124729f3b27a6e02320478a2590376b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-22 06:43:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-26 07:23:17.000000000","tz":0},"subject":"Cpy volume to image in multiple stores of glance","message":"Cpy volume to image in multiple stores of glance\n\nChange-Id: Ic7f259c3af6bd691e6d24ed180430e760d28eba8\nImplements: copy-image-in-multiple-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/992a21caa28ed0b34e6f675fdaeb3a73b52a8c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/992a21caa28ed0b34e6f675fdaeb3a73b52a8c29"}]},"branch":"refs/heads/master"},"ed007f2e00ad91bb6f6067d1aa20b1994db3b232":{"kind":"REWORK","_number":4,"created":"2019-11-26 07:29:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/695630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/695630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/4"}}},"commit":{"parents":[{"commit":"99b817c38124729f3b27a6e02320478a2590376b","subject":"Add ussuri directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/99b817c38124729f3b27a6e02320478a2590376b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-22 06:43:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-26 07:29:14.000000000","tz":0},"subject":"Cpy volume to image in multiple stores of glance","message":"Cpy volume to image in multiple stores of glance\n\nChange-Id: Ic7f259c3af6bd691e6d24ed180430e760d28eba8\nImplements: copy-image-in-multiple-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ed007f2e00ad91bb6f6067d1aa20b1994db3b232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ed007f2e00ad91bb6f6067d1aa20b1994db3b232"}]},"branch":"refs/heads/master"},"4aa49dbe53b3e653260d5985a9cd17168695ed8d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-11-26 07:53:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/695630/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/695630/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/5"}}},"commit":{"parents":[{"commit":"99b817c38124729f3b27a6e02320478a2590376b","subject":"Add ussuri directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/99b817c38124729f3b27a6e02320478a2590376b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-22 06:43:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-26 07:53:08.000000000","tz":0},"subject":"Copy volume to image in multiple stores of glance","message":"Copy volume to image in multiple stores of glance\n\nChange-Id: Ic7f259c3af6bd691e6d24ed180430e760d28eba8\nImplements: copy-image-in-multiple-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4aa49dbe53b3e653260d5985a9cd17168695ed8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4aa49dbe53b3e653260d5985a9cd17168695ed8d"}]},"branch":"refs/heads/master"},"f4254c0a77b07d511fdb9aa5576b293fbec63426":{"kind":"REWORK","_number":6,"created":"2019-12-18 06:53:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/695630/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/695630/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/695630/6"}}},"commit":{"parents":[{"commit":"e005643b639a0b734e5e716463661768f6f455ff","subject":"Merge \"[ussuri][goal] Drop python 2.7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e005643b639a0b734e5e716463661768f6f455ff"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-22 06:43:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-18 06:53:36.000000000","tz":0},"subject":"Copy volume to image in multiple stores of glance","message":"Copy volume to image in multiple stores of glance\n\nChange-Id: Ic7f259c3af6bd691e6d24ed180430e760d28eba8\nImplements: copy-image-in-multiple-stores\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f4254c0a77b07d511fdb9aa5576b293fbec63426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f4254c0a77b07d511fdb9aa5576b293fbec63426"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
