)]}'
{"id":"openstack%2Fcinder-specs~700665","triplet_id":"openstack%2Fcinder-specs~master~I78d5c360d78c65b77e84419176a68f3cf7c64db4","project":"openstack/cinder-specs","branch":"master","hashtags":[],"change_id":"I78d5c360d78c65b77e84419176a68f3cf7c64db4","subject":"update review site link","status":"MERGED","created":"2019-12-27 01:43:37.000000000","updated":"2020-01-08 14:01:12.000000000","submitted":"2020-01-08 13:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700665-1578491977475-bbef5f2b","meta_rev_id":"c01ed73e6d66ce87974aec9b5bb379844f12703f","_number":700665,"virtual_id_number":700665,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-08 13:59:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2020-01-08 13:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2020-01-08 13:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 13:53:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-08 13:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0612fe413b9bbe98bcc51609e4fc8de409e7fcc","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2019-12-27 01:43:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"934c17ce52fbbc9208784b1a733a2fe7856ae855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 01:52:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fb0bff60424c71b80a66838b100f98 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9059017f3ab4cc382dd6867ab0173ca : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"13c4b0434bf7e4b8286fae7bb0186582d977f828","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-08 13:53:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNinja-approving a minor change.","accounts_in_message":[],"_revision_number":1},{"id":"4bedb207bd5a65aa9e8b7035287a9633781b6f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 13:53:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2543f1f26fab6ecd181255ed49f3c363185229ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 13:59:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c74d267a774b57bc1e1f02ba72bf05 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d940eb33bc44e5f8a62b9d6509349b6 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"821887a0c1e80c82e541641f0a1bd107e55cd4d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 13:59:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c01ed73e6d66ce87974aec9b5bb379844f12703f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 14:01:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/bdf028438efe45f8a43ba73b5c6df509 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d00b66cb32eb974b9a1332317f644cbd1f69239","revisions":{"4d00b66cb32eb974b9a1332317f644cbd1f69239":{"kind":"REWORK","_number":1,"created":"2019-12-27 01:43:37.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/65/700665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/65/700665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/700665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/700665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/700665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/65/700665/1"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2019-12-27 01:41:47.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2019-12-27 01:43:24.000000000","tz":480},"subject":"update review site link","message":"update review site link\n\nChange-Id: I78d5c360d78c65b77e84419176a68f3cf7c64db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4d00b66cb32eb974b9a1332317f644cbd1f69239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4d00b66cb32eb974b9a1332317f644cbd1f69239"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
