)]}'
{"id":"openstack%2Fcinder-specs~700977","triplet_id":"openstack%2Fcinder-specs~master~Id6cc95b39e8ab0d295624a6ea743de9113b09753","project":"openstack/cinder-specs","branch":"master","hashtags":[],"change_id":"Id6cc95b39e8ab0d295624a6ea743de9113b09753","subject":"Add backup id to volume\u0027s metadata","status":"MERGED","created":"2020-01-03 02:16:14.000000000","updated":"2020-01-31 05:06:59.000000000","submitted":"2020-01-30 22:17:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":69,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700977-1580422637436-f323dc7b","meta_rev_id":"bc99d12d804224e77e90a8271ef822158900334a","_number":700977,"virtual_id_number":700977,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 22:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-30 13:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":2,"date":"2020-01-30 15:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2020-01-30 17:44:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-30 22:11:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 01:11:06.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2020-01-21 05:56:24.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-01-21 08:09:05.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2020-01-22 00:50:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-22 02:15:35.000000000","updated_by":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"reviewer":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"state":"REVIEWER"},{"updated":"2020-01-22 13:33:21.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-01-22 15:26:37.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-01-29 07:27:04.000000000","updated_by":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"reviewer":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"state":"REVIEWER"},{"updated":"2020-01-30 15:46:55.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-01-30 17:44:54.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-01-30 22:11:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-30 22:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9faa8f29bdd6767135d53924f997085439211030","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-03 02:16:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b82b8e857b219478ffada684f526ca27af30ee78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 02:22:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084c6a783cc8407fb365eff74d2b5240 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eebfdc65ea14d40aa7e9d7ea13ea00b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"afab0d3920f24aed9fc5de64a2abe31659663de3","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-03 02:29:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dbf9a677cb8524537daa1d804d2c743dcea7845c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-03 06:40:37.000000000","message":"Patch Set 1:\n\n(9 comments)\n\nThere are lack of some details inline.","accounts_in_message":[],"_revision_number":1},{"id":"dc9e1b3cb997b638dbe239aff0f6c53489a17b4b","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-03 06:42:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d9d3b71b11ac76365748649eeba6b1d9b69425c0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-03 06:45:27.000000000","message":"Patch Set 2: Code-Review-1\n\nPlease modify the comments in PS1.","accounts_in_message":[],"_revision_number":2},{"id":"11b14782711795fe0000b837e4cc02c6aa9ac490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 06:48:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8e158c043947189ea993ad6188a195 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a3137a074f4e87a9a7c0cc51589173 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"8b3472f46cb95f54609f5fccf11673734b4da355","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-03 10:04:57.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ff4ab691ae7323eab92468539b3aecd75974c734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 10:13:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b95b25affa4888990215f4bcfde7a5 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94c4f888bcbe4c2ebc2ff7b206ed807a : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"ae2fdde3d5def87c9584c39ccc871af793f86c95","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-03 14:50:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)\n\nThe API to create a volume from backup was only added as a convenience method to eliminate the multiple steps of creating a volume, then restoring a backup to it. I would actually not like to expand on this more by adding things like this. If anything, I think it would make sense to set the source volume ID, but even that is questionable.","accounts_in_message":[],"_revision_number":3},{"id":"debf97ea884ab1e94db95edd6f7dbc1bed7347a0","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-03 15:33:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"36cfd00ed7c8fce76ed6ef11f6ccaf1a0eb78cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 15:37:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7929996d81494ca6806f5244314819f8 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828c808ce8ee4d43b17cdd020f5d0f2b : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e34320c457dab7ecdfe16327da32dcc99862c16f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-04 01:59:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nThis will be add a new microversion, if that env does not upgrade, the request will be rejected, this should be clear.","accounts_in_message":[],"_revision_number":4},{"id":"4d1669593cc5a9eb5de1c5f6bb85ea15f0141197","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-04 02:47:59.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cdb6c0b6c190a3c64f163469d834735c8bf3be4b","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-06 00:57:11.000000000","message":"Patch Set 4:\n\n\u003e (2 comments)\n \u003e \n \u003e This will be add a new microversion, if that env does not upgrade,\n \u003e the request will be rejected, this should be clear.\n\nOK，Thanks.","accounts_in_message":[],"_revision_number":4},{"id":"abc37cc5158cd40a6a836bebfe4eb1df75db766e","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-06 00:57:33.000000000","message":"Patch Set 4:\n\n\u003e (2 comments)\n\nOK, thanks","accounts_in_message":[],"_revision_number":4},{"id":"8df3e9c52cee9b827fe86632238585daa720ead1","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-06 00:58:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"561fcc2f881e6f8543adc231687152b1129467b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 01:02:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4aced9af674567a5cdba0c72db8961 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3efab7e1a394da7af53faf0cb7aad29 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0f94c3fe8252dcbba30d4c786a4306fd96cda536","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-06 07:02:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(7 comments)\n\nLooks good, but some nits inline.","accounts_in_message":[],"_revision_number":5},{"id":"b7b731a1d2a5b0510ab3e455d22c959fb611dca5","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-06 11:01:55.000000000","message":"Patch Set 5:\n\n\u003e (8 comments)\n \u003e \n \u003e The API to create a volume from backup was only added as a\n \u003e convenience method to eliminate the multiple steps of creating a\n \u003e volume, then restoring a backup to it. I would actually not like to\n \u003e expand on this more by adding things like this. If anything, I\n \u003e think it would make sense to set the source volume ID, but even\n \u003e that is questionable.\n\nThank you for advice，I\n\n \u003e (8 comments)\n \u003e \n \u003e The API to create a volume from backup was only added as a\n \u003e convenience method to eliminate the multiple steps of creating a\n \u003e volume, then restoring a backup to it. I would actually not like to\n \u003e expand on this more by adding things like this. If anything, I\n \u003e think it would make sense to set the source volume ID, but even\n \u003e that is questionable.\n\nThank you for advice， please check again","accounts_in_message":[],"_revision_number":5},{"id":"56732c9d2dc801981a4265342891bfdd8d7a7998","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-06 11:02:13.000000000","message":"Patch Set 5:\n\n\u003e (7 comments)\n \u003e \n \u003e Looks good, but some nits inline.\n\nOK， thanks！","accounts_in_message":[],"_revision_number":5},{"id":"3a29b93d8c3ab0301aa7f8aa1d0bee196a9b30e1","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-06 11:04:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d37ab55bdbbf5938764bd6511893b2920d37d51a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 11:12:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/975a5ecacac44aedb218e477d4fe8620 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b89af4bb1bb4315b19e5a7fad090dc0 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":6},{"id":"76fad4b65e65b14253122ea28b2a8e015500a599","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-06 14:36:16.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nJust pointing out some minor things, but still not in favor of making this change. Backups don\u0027t mean anything to end users.","accounts_in_message":[],"_revision_number":6},{"id":"ed7944340289dae65c508cfaab6827822f5b635a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-07 01:38:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"be46fe54e8592b3a0cb72c636b3baad019dc4406","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-07 02:42:07.000000000","message":"Patch Set 6:\n\n\u003e (2 comments)\n \u003e \n \u003e Just pointing out some minor things, but still not in favor of\n \u003e making this change. Backups don\u0027t mean anything to end users.\n\nI think in the restored volume recorded \"backup_id\" is friendly to the end user, \nand there are some useful scenario, for example:\n1. user wants to know which backup this volume was created on\n2. For users to create again from the backup","accounts_in_message":[],"_revision_number":6},{"id":"bdb7d799d2e2a9374a84916156f2f4d8a5726be1","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-07 02:42:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"050fe58925a9975c6c7d2f3046c8c301ec6eb9b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 02:47:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d940b6aeda74caa8c2feeb48d36b6ed : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03be55e093ad4e34a054a2b4a9f3793b : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":7},{"id":"66fa4f5247b04225d7103bcb6617922b50ac2e4b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-09 01:15:02.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)\n\nSoft -1, I would not like this composing, I think on the response body we should give some note. You can refer my comments in PS1 https://review.opendev.org/#/c/700977/1/specs/untargeted/add_backup_id_to_volume.rst@54","accounts_in_message":[],"_revision_number":7},{"id":"400daf4d58fabd7740fd770cd02e20577a704972","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-09 10:05:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2fb83094ad022c4bb0c31fe223d2bb358dca7c41","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-09 10:05:55.000000000","message":"Patch Set 8: Code-Review+1\n\n\u003e (2 comments)\n \u003e \n \u003e Soft -1, I would not like this composing, I think on the response\n \u003e body we should give some note. You can refer my comments in PS1\n \u003e https://review.opendev.org/#/c/700977/1/specs/untargeted/add_backup_id_to_volume.rst@54\n\nOk, Thanks!","accounts_in_message":[],"_revision_number":8},{"id":"a307ea5d64c94e0d79bc07ed56921bbe803cd2b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:12:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c10db24e9a9465990b88fdc362015f2 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f694076da0d947b992a5fd81f36ed12f : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":8},{"id":"779d689d656274e4680091bbb3106819f9852224","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-09 11:16:06.000000000","message":"Removed Code-Review+1 by Xuan Yandong \u003cxuanyandong@inspur.com\u003e\n","accounts_in_message":[],"_revision_number":8},{"id":"0a0b36ed9e0628bfb5f1332f2ad15e0761dc2ace","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-09 11:31:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"72dccb445c4d2a2063ed6c94701094df6b252692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:39:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06b561e3ca746079358d439c35938c3 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ca6438e4f64beaa61b4c9ee2ed7ff2 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":9},{"id":"f4986d33aaefeec540f2c019181fd6c1b1d052b6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-09 12:45:27.000000000","message":"Patch Set 9: Code-Review-1\n\nPlease see discussion about this from yesterday\u0027s meeting:\n\nhttp://eavesdrop.openstack.org/meetings/cinder/2020/cinder.2020-01-08-14.00.log.html#l-143","accounts_in_message":[],"_revision_number":9},{"id":"346d61965d11a36452915b81125fd3caa3e19bfb","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-16 02:07:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fa9cf1f7829b6f8d82c67f80009d68879176ac6f","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-16 02:08:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"70de9e43b3f520818118ac16b4e85e152b2a18ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 02:16:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f945df09775485184c71fea24549585 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82daa5ccbe2e49cbabc76137b4da53a9 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":11},{"id":"fbb131df464e575725c400e38df915e4ab1e30eb","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-16 07:13:26.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9b64f62e4737a7b0794919b7f8858e5259424e6a","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-16 08:01:55.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1cd553e19af4805510c5e744fb45db2e83c372ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:10:59.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/352095d4329140d5b9783158975b44f5 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d7b4b19a64543d89ee6c9cd64442347 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":12},{"id":"b4ce567634dd6141c005b28dc5c7980c7a28b744","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-16 08:26:00.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"34e94050c4af413dde092d0fda42cd0e6d1e91fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:35:16.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55141015b6854246bd06478e249dfb98 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d990ccc6c44c4209af99d8bd28caf9cf : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":13},{"id":"edbd8d9bccf1d43dfd5e0241ba28e4a73f6c40e6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-17 23:00:55.000000000","message":"Patch Set 13: Code-Review-1\n\n(6 comments)\n\nA few requests for changes inline.  I think the proposal is fine, you just need to be a bit more specific about what you\u0027re proposing.","accounts_in_message":[],"_revision_number":13},{"id":"a16ae364913b06f43de49faacc5854ad0fc22f93","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-19 01:23:21.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"bffd08a5809b43421ff957f90a5942e748a8443f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 01:29:05.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/346912a1475845fd8dc3ba5156b4da2a : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663b736bcafc4754a30ba6c5cf4aeeaf : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":14},{"id":"f1276abdb0f3dc4a2c59afc198059e79988695e0","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-19 01:34:40.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ff1de9a405fd4dc8822a60ef1bdcda3116a61f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 01:39:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5444f64d1e7346f0a81bccd990fc41f7 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed245fa51a28481fa4d412c5478eda1b : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":15},{"id":"99a07d9b834bb4652e275b90a9b9e26595ad0275","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-20 12:58:02.000000000","message":"Patch Set 15: Code-Review-1\n\n(5 comments)\n\nAlmost there.  Noticed a typo, some formatting problems, and there\u0027s one place where I\u0027d like you to say a little more (noted inline).\n\nYou can see whether the rst file will render properly in HTML by clicking on the openstack-tox-docs link in the \u0027Zuul check\u0027 table.  It will take you to a page that has an \u0027Artifacts\u0027 section where there\u0027s a link to \u0027Docs preview site\u0027.  From there, you can find your page and make sure it looks OK.","accounts_in_message":[],"_revision_number":15},{"id":"552d4d95fc2656e1cdeb8d430528d87246e4037f","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-21 01:12:32.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"252dcd419d68205cef9a4dd4819bf96ec37fe247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 01:19:59.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/551dcad89cfa49d887090b82d6502e55 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc0918700ae422b986b0c62e6fa8d5c : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":16},{"id":"8105019821d1a6177bef588aac7dfa82709f994b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-21 02:53:39.000000000","message":"Patch Set 16: Code-Review+2\n\nLooks good!","accounts_in_message":[],"_revision_number":16},{"id":"c740678cd7548ecfca4fd0031b90fd2a4d4aa375","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-21 05:56:24.000000000","message":"Patch Set 16: Code-Review+1\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":16},{"id":"c229d10347528e86bb1f11da489ef2679c3675a4","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2020-01-21 08:09:05.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"bf7c365e2c2f59de9f6ebb84668cf1c3eb98fbf5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-22 00:50:11.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)\n\nLooks fine to me. But I wonder since the plan is now to just add a bit of metadata to an existing metadata field, a spec seems like overkill and we can just propose the code change. It also makes it seem like less of a new feature, which could be useful if we want to consider this a \"bug\" for missing information and backport it to stable branches.","accounts_in_message":[],"_revision_number":16},{"id":"5f15bedc36273523c8e4d7e5e3def6eaef43719f","author":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"date":"2020-01-22 02:15:35.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"6329443827db21b092d66b6b5658d99c57566133","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-01-22 13:33:21.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"798bd62ee3c18cba6f91b34aa8d7715d107e3156","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-01-22 15:26:37.000000000","message":"Patch Set 16: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"d3d9d429c5a89847cd82f0df9d55af41c966373d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-22 15:46:45.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)\n\nWe discussed this at today\u0027s Cinder meeting.  The general consensus was that you could follow Sean\u0027s suggestion and make this a bug instead of a feature if you like.  At the same time, it would be good to work out your answers to the questions Eric raised on this patch.","accounts_in_message":[],"_revision_number":16},{"id":"7546f33c33dceb0cb3557221549c8362ac415161","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-22 17:00:09.000000000","message":"Patch Set 16:\n\nForgot to mention that you might want to consider another name for this metadata item so that it\u0027s clear that it\u0027s an ID for the backup the volume was created from -- maybe \u0027source_backup_id\u0027 or \u0027src_backup_id\u0027, something like that.","accounts_in_message":[],"_revision_number":16},{"id":"ad3b58cd95c15041d460440302f46f54e18500b9","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-29 07:01:38.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"9ab1c0bbefe90fb9ba307b58e5c93eed58cc3c06","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-29 07:13:10.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"063be52ed8b9e38e3b54056560f1faf7ad37e4bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 07:21:30.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53d74ed467f45bd868820ff51887c12 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fccd9dd26f24f84a317df361d2fb142 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":18},{"id":"edac7c47e2a3ab30698739453ec705a27b87fdd8","author":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"date":"2020-01-29 07:27:04.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"f8ea9147b1d40d62d129731dd92e5da1186dea24","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-29 12:58:48.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)\n\nThis looks good.  Please update the \u0027Documentation\u0027 section (see comment inline) and the commit message, and I will request no more revisions (even if I see a typo on the next patch set).","accounts_in_message":[],"_revision_number":18},{"id":"dd42c45b294696f09abc86235975b9b58323bf5c","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-29 13:15:14.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"022a9d0dc93a17757d6f2aaee82db3f73e68b324","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-29 13:16:35.000000000","message":"Uploaded patch set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"056297cbc8aae0ceb182dbf4002a36dd8a6bb4e0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-29 13:25:13.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)\n\nI know I said I wouldn\u0027t request any more revisions, but there\u0027s an edit on line 119 that you really need to make!","accounts_in_message":[],"_revision_number":20},{"id":"8cab2a91238257fc49d4b3d0df8746908df6d227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 13:27:58.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29583754eb3244a18902c1019455c837 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff897211f00249d8a961c207c5a00132 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":20},{"id":"4d38e8fb041595801e907ec67884df0456d2ae36","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-29 15:16:04.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"35e4d4b5d81a2067809a0b65cd33c5022acbfd4e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-29 15:16:50.000000000","message":"Patch Set 21:\n\nSince we are so close to the spec deadline, I took the liberty of fixing the typo I noticed on the previous patch set.","accounts_in_message":[],"_revision_number":21},{"id":"30ad1b25ae911f7b87e9e31a3d1552173a996c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 15:26:02.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc75b15e552488f9947d02caf8412eb : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5e910a13ec4d16a1ce66467688ebf4 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":21},{"id":"4408251f8f7cf8a4a4a10717ad33639eff2f4d0a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-01-29 15:37:19.000000000","message":"Patch Set 21: Code-Review+2\n\nKnow that there has been some discussion that this didn\u0027t need a spec, but since we have it, it doesn\u0027t hurt to merge it.","accounts_in_message":[],"_revision_number":21},{"id":"9c0bc81cbeebb3362855fd9f13ee5d3d912df321","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-29 16:57:39.000000000","message":"Patch Set 21: Code-Review+2\n\nAgree with Jay.","accounts_in_message":[],"_revision_number":21},{"id":"0fc33ad348922caea1ad54188e67e44fecdd0dd8","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-01-30 04:49:25.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"54fb5d4d722fa49ff0572fb2f1f7965a9eb9890e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 04:55:42.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1fc6b663bb4f0d88a70f739f4fbb10 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff699277289d413c8e954b61c620963f : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":22},{"id":"cac77d9166e4397c9621a7f6d2d493f80fbc5449","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-30 13:56:08.000000000","message":"Patch Set 22: Code-Review+2\n\nAll my concerns have been addressed.","accounts_in_message":[],"_revision_number":22},{"id":"7330cacd67f6bec0a9b402ab5e6ffd36ecedec40","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-01-30 15:46:55.000000000","message":"Patch Set 22: Code-Review+2\n\nLooks good.","accounts_in_message":[],"_revision_number":22},{"id":"eab7f9c7d9fbd89250c8771d6d959842212ba430","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-30 17:44:54.000000000","message":"Patch Set 22: Code-Review+1\n\n(6 comments)\n\nI get the overall idea and i don\u0027t mind getting this in but i\u0027m not entirely convinced with the implementation.\nWhat\u0027s the purpose of making it modifiable?\nIf it\u0027s the correct response from API and important to the end user, why not make it a volume attribute like \u0027snapshot_id\u0027 or \u0027source_volid\u0027?\n\nI know i\u0027m late here and this is close to getting merged but would appreciate my queries being answered.","accounts_in_message":[],"_revision_number":22},{"id":"c5cb75ca99a980ea0213585a813d972797f293d2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-30 18:20:34.000000000","message":"Patch Set 22:\n\n\u003e What\u0027s the purpose of making it modifiable?\n\u003e If it\u0027s the correct response from API and important to the end user, why not make it a volume attribute like \u0027snapshot_id\u0027 or \u0027source_volid\u0027?\n\n(Answering for Xuan because of the timezone difference.)  See Sean\u0027s comments on earlier patch sets (concerning his skepticism that users really care about this information) and see the section \"REST API impact\" on this version.  The point isn\u0027t to make it modifiable, it\u0027s to expose some useful information to the user without having to have a new microversion every time we do this.  I think it really does belong in the volume metadata; users who value it will not mess with it, and people who don\u0027t care can delete it if it offends their sensibilities.  So it\u0027s modifiable as a side effect of being stored in the volume metadata, but I think that\u0027s ok.\n\nSuppose someone writes some kind of script that makes use of this field, and someone changes the value.  Isn\u0027t that bad?  I don\u0027t see that as different from any other use of volume metadata that end users might script for.","accounts_in_message":[],"_revision_number":22},{"id":"97a52445b10865b9d2138bba6a595661395cb421","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-30 22:11:54.000000000","message":"Patch Set 22: Workflow+1\n\nLet\u0027s go ahead and merge this.  The vocabulary clarifications that Rajat suggests can be done in a follow-up patch.  Nothing else was raised after the last call for comments was sent out:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-January/012267.html","accounts_in_message":[],"_revision_number":22},{"id":"5f186f8c29b5a454d8db5eda15a7e29511b46580","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 22:12:05.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"03a33f7c0a008c4c78c7dfc75c73560042c0ffa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 22:17:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"117e18bb56e1d56f4f1297c71fdbd7c5cf0e8139","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 22:17:17.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49a885553334c4d820649153019a4b1 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7972a892a6e844ae92035c0fc2eb9de8 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":22},{"id":"9c461b826ab2ec31e838e0b27897e960b78e76d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 22:19:00.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/0aadf1a5af3a4059b76acfdae6089d4e : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":22},{"id":"6db8cda8dcbe6a28953203d1b7d7405696157364","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-31 02:31:54.000000000","message":"Patch Set 22:\n\n(5 comments)\n\n\u003e \u003e What\u0027s the purpose of making it modifiable?\n \u003e \u003e If it\u0027s the correct response from API and important to the end\n \u003e user, why not make it a volume attribute like \u0027snapshot_id\u0027 or\n \u003e \u0027source_volid\u0027?\n \u003e \n \u003e (Answering for Xuan because of the timezone difference.)  See\n \u003e Sean\u0027s comments on earlier patch sets (concerning his skepticism\n \u003e that users really care about this information) and see the section\n \u003e \"REST API impact\" on this version.  The point isn\u0027t to make it\n \u003e modifiable, it\u0027s to expose some useful information to the user\n \u003e without having to have a new microversion every time we do this.  I\n \u003e think it really does belong in the volume metadata; users who value\n \u003e it will not mess with it, and people who don\u0027t care can delete it\n \u003e if it offends their sensibilities.  So it\u0027s modifiable as a side\n \u003e effect of being stored in the volume metadata, but I think that\u0027s\n \u003e ok.\n \u003e \n \u003e Suppose someone writes some kind of script that makes use of this\n \u003e field, and someone changes the value.  Isn\u0027t that bad?  I don\u0027t see\n \u003e that as different from any other use of volume metadata that end\n \u003e users might script for.\n\nThanks Brian and Rajat.\n\nOur initial intention was to display \"src_backup_id\" as the volume\u0027s attribute (you can see in PS9), but some Cores tend to store \"src_backup_id\" in the metadata of the volume. From the perspective of the display, it would be more intuitive to use \"src_backup_id\" as the volume attribute instead of storing \"src_backup_id\" in the metadata of the volume, but as Brian said, their role is the same, so no matter the methods are good to me.\n\nThe FUP patch is https://review.opendev.org/#/c/705146/\n\nSince we have a New Year\u0027s holiday, the reply is late.","accounts_in_message":[],"_revision_number":22},{"id":"bc99d12d804224e77e90a8271ef822158900334a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-31 05:06:59.000000000","message":"Patch Set 22:\n\nI understand the discussion more clearly now.\nif the source backup id doesn\u0027t hold much value for all the users and the integrity is less of a concern here then it seems the right way to implement it.","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"0275b38aa48f7583498dc0f1e6e9430e5fd87081","revisions":{"c2658100ff1694b18b9f17224d865d4411bb4a39":{"kind":"REWORK","_number":1,"created":"2020-01-03 02:16:14.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/1"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"subject":"Add backup_id to table volumes; When the backup restore chooses to create a new volume, write backup_id to it","message":"Add backup_id to table volumes;\nWhen the backup restore chooses to create a new volume, write backup_id to it\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c2658100ff1694b18b9f17224d865d4411bb4a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c2658100ff1694b18b9f17224d865d4411bb4a39"}]},"branch":"refs/heads/master"},"47848bc149cdbff08128f33f73eaee18a3b70eb2":{"kind":"REWORK","_number":2,"created":"2020-01-03 06:42:25.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/2"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 06:41:02.000000000","tz":480},"subject":"Add backup_id field to volumes; When the backup restore chooses to create a new volume, record \u0027backup_id\u0027 in it","message":"Add backup_id field to volumes;\nWhen the backup restore chooses to create a new volume, record \u0027backup_id\u0027 in it\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/47848bc149cdbff08128f33f73eaee18a3b70eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/47848bc149cdbff08128f33f73eaee18a3b70eb2"}]},"branch":"refs/heads/master"},"6ace01508768c0959f183debfee3e1b5caca8a78":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-03 10:04:57.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/3"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 10:00:56.000000000","tz":480},"subject":"Add backup_id field to volumes;","message":"Add backup_id field to volumes;\n\nWhen the backup restore chooses to create a new volume, record \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6ace01508768c0959f183debfee3e1b5caca8a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6ace01508768c0959f183debfee3e1b5caca8a78"}]},"branch":"refs/heads/master"},"efc7d3a9827cc77470a8226a38819c99b41fb23c":{"kind":"REWORK","_number":4,"created":"2020-01-03 15:33:10.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/4"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 15:32:03.000000000","tz":480},"subject":"Add backup_id field to volumes","message":"Add backup_id field to volumes\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/efc7d3a9827cc77470a8226a38819c99b41fb23c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/efc7d3a9827cc77470a8226a38819c99b41fb23c"}]},"branch":"refs/heads/master"},"ccae0d78ca015626b427300b3b7c9d171b6bc81f":{"kind":"REWORK","_number":5,"created":"2020-01-06 00:58:16.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/5"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-06 00:58:06.000000000","tz":480},"subject":"Add backup_id field to volumes","message":"Add backup_id field to volumes\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ccae0d78ca015626b427300b3b7c9d171b6bc81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ccae0d78ca015626b427300b3b7c9d171b6bc81f"}]},"branch":"refs/heads/master"},"954b94a5191e2cd9a139c7630c7c13348f95101c":{"kind":"REWORK","_number":6,"created":"2020-01-06 11:04:32.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/6"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-06 11:01:46.000000000","tz":480},"subject":"Add backup_id field to volumes","message":"Add backup_id field to volumes\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/954b94a5191e2cd9a139c7630c7c13348f95101c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/954b94a5191e2cd9a139c7630c7c13348f95101c"}]},"branch":"refs/heads/master"},"2397a861b6b525c3c4c159a8ce73d03968ee0e28":{"kind":"REWORK","_number":7,"created":"2020-01-07 02:42:43.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/7"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-07 02:42:29.000000000","tz":480},"subject":"Add backup_id field to volumes","message":"Add backup_id field to volumes\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2397a861b6b525c3c4c159a8ce73d03968ee0e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2397a861b6b525c3c4c159a8ce73d03968ee0e28"}]},"branch":"refs/heads/master"},"9fec1519640318168d4a506f04c27a564e473330":{"kind":"REWORK","_number":8,"created":"2020-01-09 10:05:11.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/8"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-09 10:04:58.000000000","tz":480},"subject":"Add backup_id field to volumes","message":"Add backup_id field to volumes\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9fec1519640318168d4a506f04c27a564e473330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9fec1519640318168d4a506f04c27a564e473330"}]},"branch":"refs/heads/master"},"708bb2250fcc090d64860632f4f512bd172a62f3":{"kind":"REWORK","_number":9,"created":"2020-01-09 11:31:13.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/9"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-09 11:31:01.000000000","tz":480},"subject":"Add backup_id field to volumes","message":"Add backup_id field to volumes\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/708bb2250fcc090d64860632f4f512bd172a62f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/708bb2250fcc090d64860632f4f512bd172a62f3"}]},"branch":"refs/heads/master"},"997d72ea4d91900a1e38d1f76c58fd90718b1d63":{"kind":"REWORK","_number":10,"created":"2020-01-16 02:07:19.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/10"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-16 02:06:53.000000000","tz":480},"subject":"Add backup id to volume\u0027s medadata","message":"Add backup id to volume\u0027s medadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/997d72ea4d91900a1e38d1f76c58fd90718b1d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/997d72ea4d91900a1e38d1f76c58fd90718b1d63"}]},"branch":"refs/heads/master"},"94f08e9d0b34640302ea62cedef07ddaab83806d":{"kind":"REWORK","_number":11,"created":"2020-01-16 02:08:32.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/11"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-16 02:08:21.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/94f08e9d0b34640302ea62cedef07ddaab83806d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/94f08e9d0b34640302ea62cedef07ddaab83806d"}]},"branch":"refs/heads/master"},"2fbe0f28cf6c007a01d0f20347a9aefa8c518652":{"kind":"REWORK","_number":12,"created":"2020-01-16 08:01:55.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/12"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-16 07:34:06.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2fbe0f28cf6c007a01d0f20347a9aefa8c518652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2fbe0f28cf6c007a01d0f20347a9aefa8c518652"}]},"branch":"refs/heads/master"},"1f6b68076423210756c068b0ecdceee0db5ed3c2":{"kind":"REWORK","_number":13,"created":"2020-01-16 08:26:00.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/13"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-16 08:25:50.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1f6b68076423210756c068b0ecdceee0db5ed3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1f6b68076423210756c068b0ecdceee0db5ed3c2"}]},"branch":"refs/heads/master"},"e183a7d520cde5221b302059c9f8145ad663bfaf":{"kind":"REWORK","_number":14,"created":"2020-01-19 01:23:21.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/14"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-19 01:23:05.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e183a7d520cde5221b302059c9f8145ad663bfaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e183a7d520cde5221b302059c9f8145ad663bfaf"}]},"branch":"refs/heads/master"},"c57f4eff4da59348ac543ca299029cae357ddc8e":{"kind":"REWORK","_number":15,"created":"2020-01-19 01:34:40.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/15"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-19 01:34:35.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c57f4eff4da59348ac543ca299029cae357ddc8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c57f4eff4da59348ac543ca299029cae357ddc8e"}]},"branch":"refs/heads/master"},"6e79ad7b8523ce2f5c8d0261a57498051a172516":{"kind":"REWORK","_number":16,"created":"2020-01-21 01:12:32.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/16"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-21 01:07:46.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6e79ad7b8523ce2f5c8d0261a57498051a172516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6e79ad7b8523ce2f5c8d0261a57498051a172516"}]},"branch":"refs/heads/master"},"7fd6228c98048e7fd8b56c16e4366dc3f61e55d4":{"kind":"REWORK","_number":17,"created":"2020-01-29 07:01:38.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/17"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-29 07:01:26.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7fd6228c98048e7fd8b56c16e4366dc3f61e55d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7fd6228c98048e7fd8b56c16e4366dc3f61e55d4"}]},"branch":"refs/heads/master"},"8e5cfb0363a81fa8ca7fc6e622b666059a214310":{"kind":"REWORK","_number":18,"created":"2020-01-29 07:13:10.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/18"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-29 07:13:07.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8e5cfb0363a81fa8ca7fc6e622b666059a214310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8e5cfb0363a81fa8ca7fc6e622b666059a214310"}]},"branch":"refs/heads/master"},"557e1fdf08dc6c8b84be50018ff61815fd67a77a":{"kind":"REWORK","_number":19,"created":"2020-01-29 13:15:14.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/19"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-29 13:14:47.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027src_backup_id\u0027 in it\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/557e1fdf08dc6c8b84be50018ff61815fd67a77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/557e1fdf08dc6c8b84be50018ff61815fd67a77a"}]},"branch":"refs/heads/master"},"649804778eaded41b19d2344abf5b6ada5ce8b50":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2020-01-29 13:16:35.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/20"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-29 13:15:53.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027src_backup_id\u0027 in the volume metadata.\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/649804778eaded41b19d2344abf5b6ada5ce8b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/649804778eaded41b19d2344abf5b6ada5ce8b50"}]},"branch":"refs/heads/master"},"1d29e87e9f44af056c76533779eb646d55862919":{"kind":"REWORK","_number":21,"created":"2020-01-29 15:16:04.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/77/700977/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/21"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-01-29 15:15:45.000000000","tz":0},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027src_backup_id\u0027 in the volume metadata.\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1d29e87e9f44af056c76533779eb646d55862919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1d29e87e9f44af056c76533779eb646d55862919"}]},"branch":"refs/heads/master"},"0275b38aa48f7583498dc0f1e6e9430e5fd87081":{"kind":"REWORK","_number":22,"created":"2020-01-30 04:49:25.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/700977/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/77/700977/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/77/700977/22"}}},"commit":{"parents":[{"commit":"a0d2197c01a1ce8ddf0288f61eaf379282632e20","subject":"Merge \"Adjust specs sidebar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a0d2197c01a1ce8ddf0288f61eaf379282632e20"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-03 02:07:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-01-30 04:49:09.000000000","tz":480},"subject":"Add backup id to volume\u0027s metadata","message":"Add backup id to volume\u0027s metadata\n\nWhen the backup restore chooses to create a new volume,\nrecord \u0027src_backup_id\u0027 in the volume metadata.\n\nImplements: blueprint add-volume-backup-id\n\nChange-Id: Id6cc95b39e8ab0d295624a6ea743de9113b09753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0275b38aa48f7583498dc0f1e6e9430e5fd87081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0275b38aa48f7583498dc0f1e6e9430e5fd87081"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
