)]}'
{"id":"openstack%2Fcinder-specs~712301","triplet_id":"openstack%2Fcinder-specs~master~Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5","project":"openstack/cinder-specs","branch":"master","topic":"bp/backup-backends-configuration","hashtags":[],"change_id":"Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5","subject":"Backup Backends Configuration","status":"MERGED","created":"2020-03-11 10:02:50.000000000","updated":"2023-04-12 07:07:08.000000000","submitted":"2020-07-01 16:07:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712301-1593619657321-6d9093bf","meta_rev_id":"82e6fce612a22fc8190d3b730e54b0013887432c","_number":712301,"virtual_id_number":712301,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 16:07:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2020-07-01 15:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-01 14:47:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2020-07-01 15:51:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 00:58:08.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2020-06-05 15:07:02.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-06-24 14:14:23.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-07-01 14:47:14.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-07-01 15:51:27.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-01 16:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 07:07:08.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"9e8297f4825090552328b5073fa566751b759082","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-11 10:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89d50354ccd065b9145cbac3143e136483aa2226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 10:15:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53cd54470e74903bacddd0c0cebcee7 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994be0410906460790f06e2396940ede : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e1e90501953a16088b56b1b87b1d40d7beef92ad","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-27 10:48:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65f44e49e8cd3bfe637adf11576de4cc0376be84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:53:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a937d71a5e2e431c925a5545a09ce6a7 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6a7e2d30714f9fb94599b77c30034a : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4c3a3a60b185d5e39bda850e890564fb7dbb52ed","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-05-09 00:58:08.000000000","message":"Patch Set 2: Code-Review+1\n\ni like this idea, this will be useful for users.\nLGTM.","accounts_in_message":[],"_revision_number":2},{"id":"5fcc348a76a860ea6ebccf17366e4c71c20f6cd2","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-05-09 01:07:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73da96fcdbe36992b21a65c9d8c671e11380687c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-05 15:07:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba7c5665e3603ec058cadc188cbee1e1a33ef4b3","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-22 10:19:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa5e5ab35a3bcc7b218af17d1eaa4de3e1ed29da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:29:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab8daf37bab4390ad909cb7de0273f4 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174528c1fec84235bdf915f55234d62e : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":3},{"id":"45e7371c5f621b7907745b44b626689f8b083465","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-24 14:14:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e7f2a9f8be95463227b726ce415b1640c2ab995","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-24 15:46:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nA few typos to address and the note on what we agreed to in the mid-cycle.","accounts_in_message":[],"_revision_number":3},{"id":"7af912f4b6a452934c74731d52bd45deb5deb86d","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-25 12:18:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"963f121d903e47a5b586bd55667174cd29add8c0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-25 12:21:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"256fee31500961ed1e63225b4c21105e05e6fc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 12:41:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e9e6034e2649cfa4a835e24d052038 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8bd20af93274adf8c519512c6a9817f : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"665f2b8fab3a14114a070879dc4b0d855b0dda50","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-25 21:24:14.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nOne minor nit if you push up another patch.  Otherwise looks good.","accounts_in_message":[],"_revision_number":5},{"id":"59443c075e1e8c37afb0553434b7cbb949945763","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-26 10:30:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c5fab6c2738336a7bf2b5827d2ccbb8d90bee5f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 10:40:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8aafeefd4cf400694dfbe9d6ac44041 : FAILURE in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5712c264754cb7934c81fa840b9242 : FAILURE in 5m 18s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/victoria/backup-backends-configuration.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/victoria/backup-backends-configuration.rst","accounts_in_message":[],"_revision_number":6},{"id":"6032fb3bd26e27783e5006c09d8b0e5282992662","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-26 10:58:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"af74ea1d063eb13311f6ab4d711ab54aee256bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 11:07:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0004ff204f243b5ab5a352f579a2b80 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824125a7d9344ac5952e9a0ded9c3f99 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":7},{"id":"5ffec7e7ea4668c2d40f336c71cfdebc6b9e3a32","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-28 17:10:34.000000000","message":"Patch Set 7: Code-Review+1\n\n(3 comments)\n\nThis looks good.  A few questions inline.","accounts_in_message":[],"_revision_number":7},{"id":"a6a085a0b6e654550bb98a18cbdf8acdd0263b00","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-01 09:55:37.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7cbad8bdd13e9e964a94446468071c21860df631","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-01 09:56:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5a8296c95e44c60008b4e895604b6c3f478c1560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 10:07:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b9b6a0e0fc41ce84383764634918a2 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cdf47c8a86f4a6493dae864185aa7ce : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":8},{"id":"47df2f9da6d933f01ee58ec7c504c64e51c86ece","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-07-01 14:47:14.000000000","message":"Patch Set 8: Code-Review+2\n\nChanges look good.  Thanks!","accounts_in_message":[],"_revision_number":8},{"id":"d5ece31bec08d823e3f6ea75726e9d08b5e7519f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-01 15:51:27.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nThis looks good to me, we\u0027ve discussed it at the virtual mid-cycle meeting and at today\u0027s cinder meeting, and everyone seems to be on-board with it.","accounts_in_message":[],"_revision_number":8},{"id":"e62e2a5129ce1c309955af3b08affc8884429753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 15:51:43.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"75f4b7c8904500fbd01e8d4601aba0417bed09f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 16:07:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"04d5c359ef5c3312e461a02414caef6e415e5e46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 16:07:37.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d528a7e5e3041a7b8d0d7e75802e30f : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197d491c580542158288c72f49b0a2ff : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":8},{"id":"db78fcffaec97587b7d2cc09096f4cd4934290b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 16:10:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/2f7900b9c86f46f1b2fcea2c92c9b25a : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8ffad2aec7c06070f6ca2b20baa1af3786158223","revisions":{"30607fab4cf39a9b806bed2caaae5fd383b3dc07":{"kind":"REWORK","_number":1,"created":"2020-03-11 10:02:50.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/1"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/30607fab4cf39a9b806bed2caaae5fd383b3dc07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/30607fab4cf39a9b806bed2caaae5fd383b3dc07"}]},"branch":"refs/heads/master"},"81469366f90dec9256f7170f8f3c3769059b404a":{"kind":"REWORK","_number":2,"created":"2020-04-27 10:48:01.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/2"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-27 10:47:49.000000000","tz":180},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/81469366f90dec9256f7170f8f3c3769059b404a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/81469366f90dec9256f7170f8f3c3769059b404a"}]},"branch":"refs/heads/master"},"68c7288faeed1c65baca6626997a46524734d502":{"kind":"REWORK","_number":3,"created":"2020-06-22 10:19:26.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/3"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-22 10:19:14.000000000","tz":180},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/68c7288faeed1c65baca6626997a46524734d502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/68c7288faeed1c65baca6626997a46524734d502"}]},"branch":"refs/heads/master"},"d3116d1226a06a1a37db74b0d7b36d94e8b4e52c":{"kind":"REWORK","_number":4,"created":"2020-06-25 12:18:26.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/4"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-25 12:18:13.000000000","tz":180},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d3116d1226a06a1a37db74b0d7b36d94e8b4e52c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d3116d1226a06a1a37db74b0d7b36d94e8b4e52c"}]},"branch":"refs/heads/master"},"17af3b892e7c324e83453ee7f14f56c29cd32a23":{"kind":"REWORK","_number":5,"created":"2020-06-25 12:21:15.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/5"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-25 12:21:03.000000000","tz":180},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/17af3b892e7c324e83453ee7f14f56c29cd32a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/17af3b892e7c324e83453ee7f14f56c29cd32a23"}]},"branch":"refs/heads/master"},"23e515b3e0d3753dd7f72125309d1bd28f04ee8f":{"kind":"REWORK","_number":6,"created":"2020-06-26 10:30:51.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/6"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-26 10:30:40.000000000","tz":180},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/23e515b3e0d3753dd7f72125309d1bd28f04ee8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/23e515b3e0d3753dd7f72125309d1bd28f04ee8f"}]},"branch":"refs/heads/master"},"b5d529f32c6635794455e894e03f8b5a0358d681":{"kind":"REWORK","_number":7,"created":"2020-06-26 10:58:05.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/7"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-26 10:57:56.000000000","tz":180},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b5d529f32c6635794455e894e03f8b5a0358d681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b5d529f32c6635794455e894e03f8b5a0358d681"}]},"branch":"refs/heads/master"},"8ffad2aec7c06070f6ca2b20baa1af3786158223":{"kind":"REWORK","_number":8,"created":"2020-07-01 09:56:35.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/712301/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/712301/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/712301/8"}}},"commit":{"parents":[{"commit":"3b893d223755ea09b198fdd6610b40b38ae85651","subject":"Merge \"Support volume local cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b893d223755ea09b198fdd6610b40b38ae85651"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-11 10:02:08.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-07-01 09:56:22.000000000","tz":180},"subject":"Backup Backends Configuration","message":"Backup Backends Configuration\n\nChange-Id: Ifa4c6c0ab8ef79d889a092f3caaf7304316cb3d5\nRelated: blueprint backup-backends-configuration\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8ffad2aec7c06070f6ca2b20baa1af3786158223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8ffad2aec7c06070f6ca2b20baa1af3786158223"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
