)]}'
{"id":"openstack%2Fcinder-specs~730701","triplet_id":"openstack%2Fcinder-specs~master~I622ac7f856908031c2ab1972e77c451af273d7c9","project":"openstack/cinder-specs","branch":"master","topic":"bp/remove-quota-usage-cache","hashtags":[],"change_id":"I622ac7f856908031c2ab1972e77c451af273d7c9","subject":"Remove quota usage cache","status":"ABANDONED","created":"2020-05-26 05:12:56.000000000","updated":"2021-12-07 15:31:06.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76aa469ce57c7e7fe95615342c7752844977a572","_number":730701,"virtual_id_number":730701,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-29 01:17:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"disliked":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"all":[{"value":0,"date":"2020-07-24 21:31:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":-1,"date":"2020-07-01 15:30:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":-1,"date":"2020-06-30 14:53:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":-1,"date":"2020-07-01 14:27:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":1,"date":"2020-07-01 14:29:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-27 14:07:59.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 12:47:50.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-17 12:31:03.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-06-29 01:17:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 14:53:39.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2020-07-01 14:27:31.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-07-01 14:29:32.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-07-01 15:30:43.000000000","updated_by":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"reviewer":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"state":"REVIEWER"},{"updated":"2020-07-24 21:31:53.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-10-27 14:07:59.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"fe16354e386e2245a556e70da07f53d07c2a4929","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-05-26 05:12:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99bff7f9ac2dd26580433135189701c650858a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 05:18:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c0243fb576425984e9844c8972e046 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e432b597e88f49709a01fd8432263c48 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"dbe25c5aa86e8e5cad15e225ad27cde991de9253","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-05-26 17:48:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nThanks for this spec. Please fix a couple of typos if you\u0027re gonna push another patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6a91009ca73899650ee8c81d9867ad8d495232f5","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-02 12:47:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68739dc9c86eec72df8b90bc0669606036a0813d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-25 21:40:07.000000000","message":"Patch Set 1: Code-Review+1\n\nIf this is what Nova has done, I support it.  Would like to see an answer to the performance question.  Also typos that need to be fixed.","accounts_in_message":[],"_revision_number":1},{"id":"5fb0ac29a2a84b50b9df5d91d3cb80ad718f9420","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-28 17:26:08.000000000","message":"Patch Set 1:\n\nEveryone is concerned about the performance issue.  At the Victoria Virtual PTG, someone suggested running rally with the no-op volume driver to get some performance results for computing this dynamically.\n\nAdditionally, there may still be race conditions with this approach.  But it does look promising as a way to accurate usage information.","accounts_in_message":[],"_revision_number":1},{"id":"711cf743279f8b125d74031ba3a47d684633ad75","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-28 17:26:43.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"78b405985495fdba53728160a4185b6cd6915bd8","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-06-29 01:08:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a9358a976070c00276adcbd520ea67779f2f990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 01:17:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d48683f22b4a35ab6e1315c92adb58 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6d6e10807846cfbc8d83ce214f1f62 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b1ecfd2000e9ee5d31e1c7a3eceb5c3cebabac94","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-06-30 14:53:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0a499279949121a9d85ee910801c57f2a89f2650","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-01 14:27:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15473031e579fa8aebc08dbe3947541a0ff0ec47","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-07-01 14:29:32.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for adding the performance data.  Think this is getting close.","accounts_in_message":[],"_revision_number":2},{"id":"e17edd60f54e2e810c4bde1fbdbc171d239427d6","author":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"date":"2020-07-01 15:30:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b8ddd5662001dc4aa7c8c1190ceb616ea7f7acc1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-01 16:29:43.000000000","message":"Patch Set 2:\n\nThe consensus from the mid-cycle and today\u0027s Cinder meeting is that there is support for this spec, so we\u0027d like to give you some more time to address the comments.  So this spec has a spec-freeze exception until 1600 UTC on 10 July.  Ideally, you\u0027ll have your revisions completed before next week\u0027s Cinder meeting on 8 July so we can discuss any issues at the meeting and give you time in case you need to make a final revision.","accounts_in_message":[],"_revision_number":2},{"id":"9f52345cf5a0dfc33a546be83adf31891d8a6474","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-24 21:31:53.000000000","message":"Patch Set 2: Code-Review-2\n\nThis spec missed the Victoria spec freeze.  The community is interested in the spec, however, so I encourage you to revise it and submit it to \u0027wallaby\u0027 (if you plan to work on it) or \u0027untargeted\u0027 (if you would like to see it implemented but don\u0027t plan to do it yourself).","accounts_in_message":[],"_revision_number":2},{"id":"f1346f400daf74cc0cc9d99d90ae7550a0066d58","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-10-15 15:07:15.000000000","message":"Removed Code-Review-2 by Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"76aa469ce57c7e7fe95615342c7752844977a572","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-07 15:31:06.000000000","message":"Abandoned\n\nSuperseded by https://review.opendev.org/c/openstack/cinder-specs/+/819693","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"24f4f99151c0e060fe556c1ff333b2db49d01409","revisions":{"d72272622dbeac330ac7ebf048945e1ababc27b3":{"kind":"REWORK","_number":1,"created":"2020-05-26 05:12:56.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/01/730701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/730701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/1"}}},"commit":{"parents":[{"commit":"44988971ef8734e29aff51b788ed3ebad773582f","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/44988971ef8734e29aff51b788ed3ebad773582f"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-05-26 05:11:53.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-05-26 05:12:35.000000000","tz":600},"subject":"Remove quota usage cache","message":"Remove quota usage cache\n\nImplements: blueprint remove-quota-usage-cache\n\nChange-Id: I622ac7f856908031c2ab1972e77c451af273d7c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d72272622dbeac330ac7ebf048945e1ababc27b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d72272622dbeac330ac7ebf048945e1ababc27b3"}]},"branch":"refs/heads/master"},"24f4f99151c0e060fe556c1ff333b2db49d01409":{"kind":"REWORK","_number":2,"created":"2020-06-29 01:08:31.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/01/730701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/01/730701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/01/730701/2"}}},"commit":{"parents":[{"commit":"7c195082b4609bb07ba286a1d1331942d11abce5","subject":"Correct some typos in the document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7c195082b4609bb07ba286a1d1331942d11abce5"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-05-26 05:11:53.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-06-29 01:08:19.000000000","tz":600},"subject":"Remove quota usage cache","message":"Remove quota usage cache\n\nImplements: blueprint remove-quota-usage-cache\n\nChange-Id: I622ac7f856908031c2ab1972e77c451af273d7c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/24f4f99151c0e060fe556c1ff333b2db49d01409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/24f4f99151c0e060fe556c1ff333b2db49d01409"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
