)]}'
{"id":"openstack%2Fcinder-specs~734405","triplet_id":"openstack%2Fcinder-specs~master~I8c1549d95749a5cdd8d68bae7fdc76130d937e6b","project":"openstack/cinder-specs","branch":"master","topic":"typo","hashtags":[],"change_id":"I8c1549d95749a5cdd8d68bae7fdc76130d937e6b","subject":"Correct some typos in the document","status":"MERGED","created":"2020-06-09 01:25:35.000000000","updated":"2020-06-25 21:06:00.000000000","submitted":"2020-06-25 21:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734405-1593119056862-91b529dc","meta_rev_id":"b5d4c4c64975b762a0424de0691cfc0b74dc5349","_number":734405,"virtual_id_number":734405,"owner":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 21:04:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2020-06-25 20:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2020-06-25 20:53:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 20:53:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-25 21:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a55bb3155824efabcddcde6f0865c4703909913","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-06-09 01:25:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"049989ca49bf83f31f7c8964e4db8d4caf10defc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 01:30:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a08306c48e440aab64a13716cde8b79 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27aeb147d1c84d9ab732d4400b4697f5 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"eb2e24936a2f14aca6aa7193a4c768f545714e8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-25 20:53:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNinja-approving a docs change.","accounts_in_message":[],"_revision_number":1},{"id":"4c92ad8fa51435a09d8701bc528aa77340fa89d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 20:53:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b4a628ac5dcc95c3d2f45a5ed2b143d752bfbff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 21:04:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb7410197b144cda51b2ea0f552e96d : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4713a6e1ab94085b2d5c08d5156a87f : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"90051ee4e6841bad5be960f0a5abba51aba8cbc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 21:04:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc5b678894d64395f188d3a479cf9ca0562c77c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 21:06:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/6ac848bce9b14ac4a18f8f66e8203e1e : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c195082b4609bb07ba286a1d1331942d11abce5","revisions":{"7c195082b4609bb07ba286a1d1331942d11abce5":{"kind":"REWORK","_number":1,"created":"2020-06-09 01:25:35.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/05/734405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/05/734405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/05/734405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/05/734405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/05/734405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/05/734405/1"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-06-09 01:20:51.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-06-09 01:20:51.000000000","tz":480},"subject":"Correct some typos in the document","message":"Correct some typos in the document\n\nThis patch changes \u0027Provisionning\u0027\nto \u0027Provisioning\u0027 in specs/juno/\nrestblock-driver.rst\n\nChange-Id: I8c1549d95749a5cdd8d68bae7fdc76130d937e6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7c195082b4609bb07ba286a1d1331942d11abce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7c195082b4609bb07ba286a1d1331942d11abce5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
