)]}'
{"id":"openstack%2Fcinder-specs~736111","triplet_id":"openstack%2Fcinder-specs~master~I8c26eb9f1acd43d8c780d21484c08b1e44571dde","project":"openstack/cinder-specs","branch":"master","topic":"bp/suport-revert-any-snapshot","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-08-01 13:34:03.000000000","reason":"Change was abandoned"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2024-08-01 13:34:03.000000000","reason":"Change was abandoned"},"30092":{"account":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"last_update":"2024-08-01 13:34:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8c26eb9f1acd43d8c780d21484c08b1e44571dde","subject":"Support revert any snapshot to the volume","status":"ABANDONED","created":"2020-06-17 06:39:50.000000000","updated":"2024-08-01 13:34:03.000000000","total_comment_count":97,"unresolved_comment_count":11,"has_review_started":true,"meta_rev_id":"c1bafd8cac1dafa5f3b34e81c74fa42027c29ed0","_number":736111,"virtual_id_number":736111,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-24 01:35:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":-1,"date":"2021-12-24 14:10:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 14:27:05.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-24 14:53:59.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2020-06-24 14:58:14.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-06-24 15:49:33.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-07-01 08:27:49.000000000","updated_by":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"reviewer":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"state":"REVIEWER"},{"updated":"2020-07-01 11:01:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-01 15:07:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-10-22 11:14:49.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2020-10-26 01:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-30 02:31:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-10-30 14:40:14.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-11-03 07:21:08.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-04-23 15:31:44.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2021-06-16 08:41:07.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2021-06-30 09:27:22.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-12-09 06:59:46.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-12-13 03:22:51.000000000","updated_by":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"reviewer":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"state":"REVIEWER"}],"messages":[{"id":"58a20c04f95603a650bcb1388cbc10d226c677de","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-06-17 06:39:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32d4163c2b70acc1aa38c7cb4394c56f012754b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 07:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba8a236ea7e4692a76dba23bc38a831 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7baacef2c8ad46419a735a504515b76b : FAILURE in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"21c6dd4674ffff694c02c8f9905a93d8faa4e91a","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-06-17 07:52:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"275d0b93bcd587c7c2fa3ab03136ce97b70f597a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 08:04:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c0f1b2c07b4046a0e9d77a0a288095 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ec1cc2770044ff8722be0585f0c725 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"186b26cd7765f72b66c14eeb6d0bad88fdc72768","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-24 13:48:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d47c703b99fac5ae926098144b0129ef68d99efa","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-24 14:27:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8623eaf1ed3344851cd8a918bb8d928a405bb25","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-24 14:58:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9655547e77af9c23325dd3311207b89bd55f614c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-06-24 15:18:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78cb7b66d04ff4c496e9f34d7c19f7d04de6d39d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-24 15:49:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nFound a typo.  Also there are a lot more details required here based on mid-cycle discussion.","accounts_in_message":[],"_revision_number":2},{"id":"fc526387312c3a235069819b78efedfb01c4faa6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-25 11:59:56.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nComment inline.","accounts_in_message":[],"_revision_number":2},{"id":"12fa521572772bdacb7209d58678c05bc4bad5f9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-06-26 00:43:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b596711879769f4252170580041618631f57c32","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-01 02:56:23.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\nThanks for review, and give help。I think that\nIf the temporary volume is created, then in the case of a relatively large volume, the storage itself will be wasted and the snapshot rollback time will be increased, because the process of copying data is completed on the openstack node, and the storage fast copy feature cannot be used. so the effect will be greater in efficiency. \n\nSo I think that if the storage vendor does not support the rollback of middle snapshots, it is limited","accounts_in_message":[],"_revision_number":2},{"id":"85da916c2b20f3de1f6735957d13c92747aaf53b","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-01 02:57:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"64ec7996d59da7c18ae45167d16f9ae95e377aa0","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-01 02:58:52.000000000","message":"Patch Set 2:\n\n\u003e (2 comments)\n\nI got it, thanks very much, I described it in detail in the proposed change，","accounts_in_message":[],"_revision_number":2},{"id":"ec6f4f13ffe3da94f323c13f2c56b7d025edd95e","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-01 02:59:25.000000000","message":"Patch Set 2:\n\n\u003e (2 comments)\n\nGot it，thanks very much","accounts_in_message":[],"_revision_number":2},{"id":"786d87cff476728e9ae89e5124d7667cd1c1fa8a","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-01 02:59:45.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nGot it，thanks very much","accounts_in_message":[],"_revision_number":2},{"id":"03087249f59c962239cd5a29c145ed43656ae029","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-01 03:00:03.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Found a typo.  Also there are a lot more details required here\n \u003e based on mid-cycle discussion.\n\nGot it，thanks very much","accounts_in_message":[],"_revision_number":2},{"id":"58a742070ffd3851e581da75439f209183c23a66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 03:04:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8de45c1a9f84fb5a9701c2e87a5a7be : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d5787641a8e459d85fb041df10c318e : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"d4854e5a6f247af09ca6a29c5a6e42b1584801cd","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-01 06:07:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"46bdcf835e8817b0031eb8793e95e04d5b5ca00f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 06:20:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124ee40e0d3b415c8e65216f1a7c89c5 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2599df768ac745328d24b51b8ee95b50 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c936dd3595a32299fe68d1eb9910253a6a3d139e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-01 11:01:57.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"775dcbf23c2630d96daef7edbc2771c48380efdd","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-01 15:07:26.000000000","message":"Patch Set 4: Code-Review-1\n\nThis spec doesn\u0027t really spell out the details and implications of how this works...\n\nIf you revert to an older snapshots, are the newer snapshots all still fully functional on the backend?  Can you revert to an older snapshot and then revert to a newer snapshot?  How do we know what works on each backend?\n\nAre we going to support a generic mechanism to copy data around for backends that can\u0027t natively do this?\n\nWe may want a config option for deployments to not allow reverting to older snapshots if the operator decides that it has excessive performance overhead.","accounts_in_message":[],"_revision_number":4},{"id":"d69e5b0d8d900bf8b466e3d24cb4ff0bc81473c5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-01 18:32:28.000000000","message":"Patch Set 4: Code-Review-2\n\nWe are now in the spec freeze for Victoria.  You may continue to work on this spec for Wallaby.\n\nIn addition to the comments on this review, it would be good if you could give some details on what happens on a typical backend that can do this in the ideal way.  What happens to the more recent snapshots if you revert to an older snapshot?","accounts_in_message":[],"_revision_number":4},{"id":"79998a46a40dd41e54ddd54c2ee4ef54ae9d0401","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-16 01:32:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"02f251174b87acf84bac4cfc16381f877923a4b5","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-16 01:32:59.000000000","message":"Patch Set 4:\n\n\u003e (5 comments)\n\nok， thanks very much， please recheck.","accounts_in_message":[],"_revision_number":4},{"id":"7d2c0e2194336c0990b38bae2a47ae4b360b0af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 01:37:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d12f48a6f648d0992d8e6b52cb3e18 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c20f08590c4134969e2fad62f6a379 : FAILURE in 4m 38s","accounts_in_message":[],"_revision_number":5},{"id":"dd3e7f4789b4f34b126645d653db5dea049fb09b","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-07-16 01:45:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3604aa43048eaa90715cd23e2832a3650fefb683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 01:55:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7507f0aa4b44d0a925697faaa79f53 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6a0ba583a842699e504bffccafac15 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":6},{"id":"d4dab3bc5ac70aeeafd0b037076b1120e521aeef","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-12 03:06:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"98730aff9141cf0b2fbe3f87473990da1eed2d9d","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-12 03:08:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6e462642fe899f67c6fc3abc032437d98afb97c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 03:16:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f77f779964f4ed1aacc4d98771f24ad : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b685f63aaac14691a2098e9f631d785c : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":8},{"id":"c207eddd01720beeb982daaf2f0fcd91dac3b5e8","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-10-22 11:14:49.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nI really like the idea of having this feature :-)","accounts_in_message":[],"_revision_number":8},{"id":"6d042d434cb18e58a9229955f3697de81a8448b8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-10-22 12:06:56.000000000","message":"Patch Set 8: -Code-Review\n\nDropping -2 as this is correctly targeted for Wallaby.","accounts_in_message":[],"_revision_number":8},{"id":"61cb94a60036d5938e6c3e9baff1906f32220045","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-10-22 12:19:26.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nFormatting issue noted inline.","accounts_in_message":[],"_revision_number":8},{"id":"885bffe0b7d4172885245a12e3351baa976a5bdc","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-26 01:18:47.000000000","message":"Patch Set 8:\n\n\u003e (1 comment)\n \u003e \n \u003e I really like the idea of having this feature :-)\n\nThanks very much，good suggestion","accounts_in_message":[],"_revision_number":8},{"id":"eb721818f824c0c976f49b3cc6c4335ab0e381a7","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-26 01:21:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"013aa3aeafcd50eefafb07f1805b73452c6fe81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 01:26:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f28cbe5890489cb884d4ec9f0e929c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1504bb1a38f441b9f42ef927f0529f3 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":9},{"id":"f22d5bb8284b464e92baa4c0079270b4a809891c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-10-30 02:31:39.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)\n\nOverall, I support this, but there are a few things I\u0027d like you to address inline.","accounts_in_message":[],"_revision_number":9},{"id":"0b9124799055a640b50c8efbfc1f965b11bee37d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-10-30 14:40:14.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e614ea3c4196b2725848c375fa9fa36c90ef05ec","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-03 07:21:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"55608fb450e82b98e34ea799563342968f4db7b3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-10 02:57:59.000000000","message":"Patch Set 9:\n\nWe discussed this spec at the Wallaby PTG.  Please address the points outlined in the summary:\nhttps://wiki.openstack.org/wiki/CinderWallabyPTGSummary#Support_revert_any_snapshot_to_the_volume\n\nThanks!","accounts_in_message":[],"_revision_number":9},{"id":"0d5ecc45ee8072fd57f8793dbdde7aebe8549589","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-11-25 02:58:53.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9:\n\u003e \n\u003e We discussed this spec at the Wallaby PTG.  Please address the points outlined in the summary:\n\u003e https://wiki.openstack.org/wiki/CinderWallabyPTGSummary#Support_revert_any_snapshot_to_the_volume\n\u003e \n\u003e Thanks!\nOk, about the discussion in PTG. I have the following thoughts\n1. About how to handle the performance problems with the generic，We can set a switch, which is turned on by default\n2. About the revert chain, if the 3 is reverting, in the storage management, they can not revert to 7;\n3. I think not all the storage vendors design snapshots as tree, perhaps the internal tree, but there are no restrictions on operation. Like this, there are three snapshots of volume: snap1, snap2, snap3, we can delete snap1 in the storage management.\n4. It is possible to design an interface to realize whether the revert operation is supported by the driver","accounts_in_message":[],"_revision_number":9},{"id":"45be1d366c9dfd90723643d5d661066110639854","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-12-01 01:08:39.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"c02ab5637374da6677430d27c9f0cb4a8086c950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 01:13:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1c52aed7d542619e9f73c5817cdbe5 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9bbfbb2f831440198375867e8c77e57 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":9},{"id":"70002fc39af05d1fa3d71a8a23ec1768acaa9e73","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-14 21:38:08.000000000","message":"Patch Set 9:\n\nPlease incorporate your comment on the review into the text of the spec.  Thanks!","accounts_in_message":[],"_revision_number":9},{"id":"ad54c66a8370ed4c8e73bce2502155949be82e41","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-14 21:40:41.000000000","message":"Patch Set 9:\n\nAlso, I think Luigi\u0027s point wasn\u0027t that you should not add functional tests.  Rather, you need tests that can be run against a real storage backend, so you will probably want to add tests to the cinder-tempest-plugin (which I think would be a good idea).","accounts_in_message":[],"_revision_number":9},{"id":"8ddad1ac92438b17eb646eb671b4ff2843b59d6b","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-12-25 01:29:33.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9:\n\u003e \n\u003e Please incorporate your comment on the review into the text of the spec.  Thanks!\n\nOK, thanks very much","accounts_in_message":[],"_revision_number":9},{"id":"f20c948b2a42721c6c6d29971bbe584e9dc9fe84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-12-25 01:36:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0b16d4c42a1c82dce3264fc2f5740ad71feef986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 01:41:11.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3b1e4d81fc490786445edf65b10a43 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0250fcaae94ba6a8d717ee7ed52d20 : FAILURE in 2m 45s","accounts_in_message":[],"_revision_number":10},{"id":"8c6908de6f736302ce2a19bb09964987646734e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-12-25 01:45:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9710bcebb1017bca9ae03b34eff45820771ab077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 01:54:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/182228c0dc5945738fbcab69b590c83c : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f5ff1f57ba4ec793ef694145810920 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":11},{"id":"8aa69b2744efc890caa7820092949e7107b3d3fa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-30 09:43:42.000000000","message":"Topic set to bp/suport-revert-any-snapshot","accounts_in_message":[],"_revision_number":11},{"id":"b90a77b8df4edb67032c901f6db2bd4954b133e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-11 01:09:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2de6273df8b58ba316c0b485f64636c1d9b81ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 01:17:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d5280f11aec484db05112eaefa6a39c : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb08b6e006bf4e54a7631c22254cde26 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":12},{"id":"a025da37ca05d9b0c8d133cb33f2dbaf61ae8247","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-04-23 15:31:44.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"31f9069504da94cc815006880cdb46a1e7d3507f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-04-25 06:26:24.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f7e58ae369f109c59e010d270ff6b070ac62cb02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 06:39:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475ccf8c48814f4ab87023e994b5771a : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ddaa625ab9c4258ad5214150b7a9c46 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":13},{"id":"e92bcd3b14223bc9150a8bb7ff566e99c8f84b6e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-04-26 14:13:32.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"e154c819ccae82be9bb72c3245f7bc24a788c421","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-04 13:14:11.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)\n\nPlease see the team\u0027s discussion of this spec at the R-18 midcycle:\nhttps://wiki.openstack.org/wiki/CinderXenaMidCycleSummary#Support_revert_any_snapshot_to_the_volume","accounts_in_message":[],"_revision_number":13},{"id":"f8a86e9cc0b06259120b0f11ef1da02444da4426","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2021-06-16 08:55:57.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"dd08177438e1559425fc3898cf833e21e8c333b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-28 06:37:46.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"6a522a7f6d1a7243f2191a280155396142aaf814","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-28 06:40:21.000000000","message":"Patch Set 13:\n\n(3 comments)\n\n\u003e Patch Set 13: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Please see the team\u0027s discussion of this spec at the R-18 midcycle:\n\u003e https://wiki.openstack.org/wiki/CinderXenaMidCycleSummary#Support_revert_any_snapshot_to_the_volume\n\nOk, got it, please review it again, thanks very much.","accounts_in_message":[],"_revision_number":13},{"id":"c382bae8e103ec766312075928fdc60aec9be084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 06:50:01.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5c0fc2bbfa4d63bed93a0ca408c700 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c0e909cbff45c893a5af0733562acf : FAILURE in 7m 17s","accounts_in_message":[],"_revision_number":14},{"id":"2910947546b6026a54b860616c020cb105b39729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-29 01:58:51.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"f8ac7927cb89253c3f8b3c4576ba9f951e48c5d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:22:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3616fc74fab64607ac96eebfa040c611 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45281a1e99346acb8a3bf3a666ac507 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":15},{"id":"73cf9c71e22e5de4f7a565e1dccc4b51681df868","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-29 12:58:01.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"c0cda4f4039c8af8b580a3c04d8285eae6225c2e","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-06-30 09:27:22.000000000","message":"Patch Set 15: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"b219c66e4e8b985907b82533fa05b5a287136b09","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-06-30 11:38:20.000000000","message":"Patch Set 15: Code-Review-1\n\n(9 comments)\n\nThank you for all the improvements you\u0027ve been doing to the spec, it is looking a lot better than the last time I reviewed it.\n\nI have added some comment with some nits (which you are free to ignore) and some other things that I think need to be changed.","accounts_in_message":[],"_revision_number":15},{"id":"e01fed4ab6a6078a88503473f8aa50664d19e716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-07-02 06:26:25.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c9f4378f8e3d2e031266aca7a0f7be3042ca8aaf","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-07-02 06:28:03.000000000","message":"Patch Set 16:\n\n(9 comments)\n\nThank you very much for such a detailed review.\nI\u0027ve revised it, please review again.","accounts_in_message":[],"_revision_number":16},{"id":"021c3ee30887b6cce457881106d28d3df8f349ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 06:38:12.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275b22b0eb154b48940d45737bab7188 : FAILURE in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b178e27d7c465b882486ed76718f6d : FAILURE in 4m 32s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/xena/support-revert-any-snapshot.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/xena/support-revert-any-snapshot.rst","accounts_in_message":[],"_revision_number":16},{"id":"c571d3e21d1b51a84659ede7b327a37aa88276b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-07-06 06:31:20.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"62e307bcb9244b75c14ede9f0e3a67217f5287d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 06:40:17.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e27a36c20e24808b3931e62d4555977 : FAILURE in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0c7883dcdc4d25b2af422f2793d95b : FAILURE in 4m 55s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/xena/support-revert-any-snapshot.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/xena/support-revert-any-snapshot.rst","accounts_in_message":[],"_revision_number":17},{"id":"2bafa02c2cd9bb9d28a9e1aec71a6a116e86d8f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-07-06 11:24:27.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9f053f10cb1cdd3f0c7de572e2222c900015df31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 11:30:41.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4436496ee63f45ee84107341e7c1f46e : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e703a69178b4e43abbe5101bacb47ed : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":18},{"id":"aa4bbdf735a0b04180aa5947cc448f7bfd2961b0","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-07-06 13:36:48.000000000","message":"Patch Set 18: Code-Review-1\n\n(3 comments)\n\nThis is getting there - just a few things...","accounts_in_message":[],"_revision_number":18},{"id":"bf36b107991038d1568ebb3391866243a79421ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-07-07 03:21:48.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"7f2285655444722587bdd115cc4de7a68b823c60","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-07-07 03:23:03.000000000","message":"Patch Set 19:\n\n(3 comments)\n\nOk, thanks very much for your review.","accounts_in_message":[],"_revision_number":19},{"id":"884ad16ea3f96c2da8974e522fa87911014b7ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 03:28:04.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2accd21f784344a973a85467539c10 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edad975fb2f24956a1d77e9c4227a360 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":19},{"id":"22b0357a7ae7d992cddfb7dbbff83f656dab236b","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-07-07 12:44:26.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"c0b2c8703111b2952e15eb0939d26f0cb0582566","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-08 09:38:54.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"958070a82f72767e1f02ab17e57d16a3ff84005e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-09-09 00:45:29.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"fffab96822a624f92e838d88eb586905625fc4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 01:07:17.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c592ef41664631bb9c19d44a27aa88 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e782394943194300934ca084729fbde9 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":20},{"id":"e55dfb14fc73606b7351a7809dbf26b3104b9d76","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-21 07:15:29.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"cf820d64b9252b15aab83142d46cc591996cccf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-10-19 01:28:53.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"390521275a3e334ae1028b695b66241a45bc99bd","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-10-19 01:30:52.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"df60658ff1d2b3829e3432a3ec8aa48c73176964","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-10-19 01:31:10.000000000","message":"Patch Set 21: -Code-Review","accounts_in_message":[],"_revision_number":21},{"id":"7e34c283612f54093e4ebaaeba8ee95dfb15ffcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 01:39:44.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1296c0a3241849cfb0478b1a6f742cae : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2428df28e045adaf875ebd78133f30 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":21},{"id":"596a0a7e2038caf79a1ddc1e94cf3efd86fb990a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-11 15:45:50.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"fbb8bdb3e3a9fa9d949c8e581bc480c6fb0f414d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-11 15:48:50.000000000","message":"Patch Set 21: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"7f9088e80dd4a85e7be19188d8f82be29ac16606","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-09 06:59:46.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d9c44832141b728d6bcd764405bd720fe68704bb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-09 13:37:14.000000000","message":"Patch Set 21: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":21},{"id":"af346b827a197afe853f016cb1c005d922d21c9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-12-10 02:37:05.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"49b226bc5d2c558f7fe1669e4718a59f5d6ac8cc","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-12-10 02:37:29.000000000","message":"Patch Set 22:\n\n(7 comments)","accounts_in_message":[],"_revision_number":22},{"id":"3344c26841885a2c5b9dbc8986d8be82bc95aa44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 02:55:20.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e952e3ad584f71852258b04dd53f8d : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5da0ea0afc4953b20a297bfc6d9f07 : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":22},{"id":"05a5c0efe9ec67f18f5b75ada7f1f40f31dcaf29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-12-10 03:00:25.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"b00ec401694c214d4b69c1dd201e492c7b838063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 03:16:17.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65da1638de74d92bba0e153e7b6289d : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8cf74491e4e4729bcd8b53d4ea297d3 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":23},{"id":"dbf47ae600b8f64d75396cbf66901836e8f6187a","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-12-13 03:22:51.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"77377e4e3cd47028d348a2c42b3fd515d64a76a3","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-12-13 04:52:39.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"2654ae39c315510d4bc0eda6a9f49c16f86b99da","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-12-13 05:37:13.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"bafc95bc0ee27d8da84ab1487de4dd27683881d9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-19 18:06:48.000000000","message":"Patch Set 23: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":23},{"id":"7f96e1da6d7744ef5ada22dfc6e502988a7d3ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-12-24 01:29:50.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"2a56460577ced2793a30e37abcba3ced3d84a601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 01:35:37.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a301622e0024401be4a20385c749d1a : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b429466a21ae4ee4bee8fa7650cbfc07 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":24},{"id":"a2f43dac1f674fee8ebae982081ccb535edab04a","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-12-24 02:08:08.000000000","message":"Patch Set 24:\n\n(5 comments)","accounts_in_message":[],"_revision_number":24},{"id":"d8eb09d1d6db8b8560543cdcb49109052b6338a9","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-12-24 14:10:13.000000000","message":"Patch Set 24: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"fb0c261f7187ac4382e87c59423fd86f6e2bf238","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-01-06 20:27:22.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"1a7f42b8f541d5de08708195eb7910e82bec0966","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-03-03 17:39:27.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"c1bafd8cac1dafa5f3b34e81c74fa42027c29ed0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-01 13:34:03.000000000","message":"Abandoned\n\nNot updated since Yoga","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"bdc17270c80b337bf1914a3c3c26a8bceaf81c14","revisions":{"3d5fab962e7f27da19aa6bfcfce1c006423e3490":{"kind":"REWORK","_number":1,"created":"2020-06-17 06:39:50.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/1"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:26.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3d5fab962e7f27da19aa6bfcfce1c006423e3490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3d5fab962e7f27da19aa6bfcfce1c006423e3490"}]},"branch":"refs/heads/master"},"0377dd0ca9d46c0dae50df0c6e7790f962d415b4":{"kind":"REWORK","_number":2,"created":"2020-06-17 07:52:08.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/2"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 07:51:57.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0377dd0ca9d46c0dae50df0c6e7790f962d415b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0377dd0ca9d46c0dae50df0c6e7790f962d415b4"}]},"branch":"refs/heads/master"},"571d7c5005af361012749cd9ed2688d68a986f03":{"kind":"REWORK","_number":3,"created":"2020-07-01 02:57:21.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/3"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-07-01 02:56:53.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/571d7c5005af361012749cd9ed2688d68a986f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/571d7c5005af361012749cd9ed2688d68a986f03"}]},"branch":"refs/heads/master"},"dbca0bcbf96d2f571c19061f061d7ef09ac44031":{"kind":"REWORK","_number":4,"created":"2020-07-01 06:07:02.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/4"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-07-01 06:06:54.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/dbca0bcbf96d2f571c19061f061d7ef09ac44031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/dbca0bcbf96d2f571c19061f061d7ef09ac44031"}]},"branch":"refs/heads/master"},"5086ba33d3c32cc5b15d442df2502667672264ac":{"kind":"REWORK","_number":5,"created":"2020-07-16 01:32:19.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/5"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-07-16 01:32:03.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5086ba33d3c32cc5b15d442df2502667672264ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5086ba33d3c32cc5b15d442df2502667672264ac"}]},"branch":"refs/heads/master"},"1bc7aad4cb52d8b6b9fbe2a039c7f7ac850d14c7":{"kind":"REWORK","_number":6,"created":"2020-07-16 01:45:39.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/6"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-07-16 01:45:27.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1bc7aad4cb52d8b6b9fbe2a039c7f7ac850d14c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1bc7aad4cb52d8b6b9fbe2a039c7f7ac850d14c7"}]},"branch":"refs/heads/master"},"792e2dfff3d3fbbcd26591183f5c66e209c49afe":{"kind":"REWORK","_number":7,"created":"2020-10-12 03:06:06.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/7"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-10-12 03:05:47.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/792e2dfff3d3fbbcd26591183f5c66e209c49afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/792e2dfff3d3fbbcd26591183f5c66e209c49afe"}]},"branch":"refs/heads/master"},"09ae3c6180ba1c4eae071a6f8606a3e633799770":{"kind":"REWORK","_number":8,"created":"2020-10-12 03:08:11.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/8"}}},"commit":{"parents":[{"commit":"00c573d0d96fcbb2311aa03e540c7b38775a7cad","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/00c573d0d96fcbb2311aa03e540c7b38775a7cad"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-10-12 03:08:00.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/09ae3c6180ba1c4eae071a6f8606a3e633799770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/09ae3c6180ba1c4eae071a6f8606a3e633799770"}]},"branch":"refs/heads/master"},"1025808c70fbd186b550e1bfe350c82cb4a57d69":{"kind":"REWORK","_number":9,"created":"2020-10-26 01:21:15.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/9"}}},"commit":{"parents":[{"commit":"873135ff0529e57a1177f00def79b01c167ce740","subject":"Merge \"Retarget specs for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/873135ff0529e57a1177f00def79b01c167ce740"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-10-26 01:20:57.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1025808c70fbd186b550e1bfe350c82cb4a57d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1025808c70fbd186b550e1bfe350c82cb4a57d69"}]},"branch":"refs/heads/master"},"dc4d549b1e59cf6f1fcade43cc95a7209ef7b85d":{"kind":"REWORK","_number":10,"created":"2020-12-25 01:36:44.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/10"}}},"commit":{"parents":[{"commit":"873135ff0529e57a1177f00def79b01c167ce740","subject":"Merge \"Retarget specs for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/873135ff0529e57a1177f00def79b01c167ce740"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yadiel Xuan(轩艳东)","email":"xuanyandong@inspur.com","date":"2020-12-25 01:30:15.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/dc4d549b1e59cf6f1fcade43cc95a7209ef7b85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/dc4d549b1e59cf6f1fcade43cc95a7209ef7b85d"}]},"branch":"refs/heads/master"},"1e57fd01b801fb2568ab52f1a890ec5796622408":{"kind":"REWORK","_number":11,"created":"2020-12-25 01:45:28.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/11"}}},"commit":{"parents":[{"commit":"873135ff0529e57a1177f00def79b01c167ce740","subject":"Merge \"Retarget specs for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/873135ff0529e57a1177f00def79b01c167ce740"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yadiel Xuan(轩艳东)","email":"xuanyandong@inspur.com","date":"2020-12-25 01:45:13.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1e57fd01b801fb2568ab52f1a890ec5796622408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1e57fd01b801fb2568ab52f1a890ec5796622408"}]},"branch":"refs/heads/master"},"a3175954112fb1c4d49225fd21b44c559259eaad":{"kind":"REWORK","_number":12,"created":"2021-03-11 01:09:41.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/12"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-03-11 01:09:19.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a3175954112fb1c4d49225fd21b44c559259eaad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a3175954112fb1c4d49225fd21b44c559259eaad"}]},"branch":"refs/heads/master"},"9accb245b5c7c3c9f9aca40cf96f289dca4c4e7f":{"kind":"REWORK","_number":13,"created":"2021-04-25 06:26:24.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/13"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-04-25 06:26:14.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9accb245b5c7c3c9f9aca40cf96f289dca4c4e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9accb245b5c7c3c9f9aca40cf96f289dca4c4e7f"}]},"branch":"refs/heads/master"},"9097643585e21e2acacb7aeeca015a0f0d33fad5":{"kind":"REWORK","_number":14,"created":"2021-06-28 06:37:46.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/14"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-06-28 06:37:19.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9097643585e21e2acacb7aeeca015a0f0d33fad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9097643585e21e2acacb7aeeca015a0f0d33fad5"}]},"branch":"refs/heads/master"},"4b93ec863b0a3d725532c21e3e6295d915f82696":{"kind":"REWORK","_number":15,"created":"2021-06-29 01:58:51.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/15"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-06-29 01:58:40.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4b93ec863b0a3d725532c21e3e6295d915f82696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4b93ec863b0a3d725532c21e3e6295d915f82696"}]},"branch":"refs/heads/master"},"40b6591f275ea8343a0f9ee24fb48df9e94fe967":{"kind":"REWORK","_number":16,"created":"2021-07-02 06:26:25.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/16"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-07-02 06:26:08.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/40b6591f275ea8343a0f9ee24fb48df9e94fe967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/40b6591f275ea8343a0f9ee24fb48df9e94fe967"}]},"branch":"refs/heads/master"},"5a35c7bd1c182e6b90d3548bbf8e2fed31453321":{"kind":"REWORK","_number":17,"created":"2021-07-06 06:31:20.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/17"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-07-06 06:31:04.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5a35c7bd1c182e6b90d3548bbf8e2fed31453321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5a35c7bd1c182e6b90d3548bbf8e2fed31453321"}]},"branch":"refs/heads/master"},"408d008e39ef49e38f361a76a4ce3d46da371d8c":{"kind":"REWORK","_number":18,"created":"2021-07-06 11:24:27.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/18"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-07-06 11:24:18.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/408d008e39ef49e38f361a76a4ce3d46da371d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/408d008e39ef49e38f361a76a4ce3d46da371d8c"}]},"branch":"refs/heads/master"},"f6f4b1991c189145ff7c151fa12f3721dfd26e4a":{"kind":"REWORK","_number":19,"created":"2021-07-07 03:21:48.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/19"}}},"commit":{"parents":[{"commit":"14f350682243588c730202fcc566717c0730c6e1","subject":"Merge \"Remove testr detrius\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/14f350682243588c730202fcc566717c0730c6e1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-07-07 03:21:39.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f6f4b1991c189145ff7c151fa12f3721dfd26e4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f6f4b1991c189145ff7c151fa12f3721dfd26e4a"}]},"branch":"refs/heads/master"},"7873ce47e23fadb32f32018259f6aa9c32ee4efb":{"kind":"REWORK","_number":20,"created":"2021-09-09 00:45:29.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/20"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-06-17 06:38:20.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-09-09 00:45:09.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7873ce47e23fadb32f32018259f6aa9c32ee4efb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7873ce47e23fadb32f32018259f6aa9c32ee4efb"}]},"branch":"refs/heads/master"},"6b422dd91b1070cee65e63cbfdb40bb7ba4143e5":{"kind":"REWORK","_number":21,"created":"2021-10-19 01:28:53.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/21"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"xuanyd","email":"xuanyandong@inspur.com","date":"2021-10-19 01:28:40.000000000","tz":480},"committer":{"name":"xuanyd","email":"xuanyandong@inspur.com","date":"2021-10-19 01:28:40.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6b422dd91b1070cee65e63cbfdb40bb7ba4143e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6b422dd91b1070cee65e63cbfdb40bb7ba4143e5"}]},"branch":"refs/heads/master"},"d3bb40ed67609dfc537503f29b11ea7afbf89591":{"kind":"REWORK","_number":22,"created":"2021-12-10 02:37:05.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/22"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"xuanyd","email":"xuanyandong@inspur.com","date":"2021-10-19 01:28:40.000000000","tz":480},"committer":{"name":"xuanyd","email":"xuanyandong@inspur.com","date":"2021-12-10 02:36:43.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d3bb40ed67609dfc537503f29b11ea7afbf89591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d3bb40ed67609dfc537503f29b11ea7afbf89591"}]},"branch":"refs/heads/master"},"eb166a072b00a97bb3069a00179265c8f5155e37":{"kind":"REWORK","_number":23,"created":"2021-12-10 03:00:25.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/23"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"xuanyd","email":"xuanyandong@inspur.com","date":"2021-10-19 01:28:40.000000000","tz":480},"committer":{"name":"xuanyd","email":"xuanyandong@inspur.com","date":"2021-12-10 03:00:13.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/eb166a072b00a97bb3069a00179265c8f5155e37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/eb166a072b00a97bb3069a00179265c8f5155e37"}]},"branch":"refs/heads/master"},"bdc17270c80b337bf1914a3c3c26a8bceaf81c14":{"kind":"REWORK","_number":24,"created":"2021-12-24 01:29:50.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/11/736111/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/11/736111/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/11/736111/24"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"xuanyd","email":"xuanyandong@inspur.com","date":"2021-10-19 01:28:40.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-12-24 01:29:36.000000000","tz":480},"subject":"Support revert any snapshot to the volume","message":"Support revert any snapshot to the volume\n\nThis feature will support revert other snapshot\nin the snapshot chain to the volume, not only\nthe latest one.\n\nChange-Id: I8c26eb9f1acd43d8c780d21484c08b1e44571dde\nImplements: blueprint suport-revert-any-snapshot\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bdc17270c80b337bf1914a3c3c26a8bceaf81c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bdc17270c80b337bf1914a3c3c26a8bceaf81c14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
