)]}'
{"id":"openstack%2Fcinder-specs~743084","triplet_id":"openstack%2Fcinder-specs~master~I482ac98bc56f0e3cfb8b767f47649da11ed1afab","project":"openstack/cinder-specs","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I482ac98bc56f0e3cfb8b767f47649da11ed1afab","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-07-26 22:53:00.000000000","updated":"2020-08-02 23:57:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"da2c8443c9ed971413a21afff19d4381d9e8a9b2","_number":743084,"virtual_id_number":743084,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-26 23:06:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-26 23:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"112f109877514aad38f114c9e542b2181cb54a9c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-26 22:53:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5195420d57ffaf2e71076db3ade80f79bbf5d3f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 23:06:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19fef652be2744229e7f566f6f028731 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5520f57f70d74ce8bb21926dc329923b : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a821a08a7d7edd134b0f151049ecd4d561255d11","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:57:45.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b83c40484ad9c703bd93d9e33c5dad5f724fe1e","revisions":{"5b83c40484ad9c703bd93d9e33c5dad5f724fe1e":{"kind":"REWORK","_number":1,"created":"2020-07-26 22:53:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/743084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/84/743084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/84/743084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/84/743084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/84/743084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/84/743084/1"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 22:49:43.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 22:52:58.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional tests jobs\n- docs jobs\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\nStory: #2007865\nTask: #40179\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5b83c40484ad9c703bd93d9e33c5dad5f724fe1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5b83c40484ad9c703bd93d9e33c5dad5f724fe1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
