)]}'
{"id":"openstack%2Fcinder-specs~749991","triplet_id":"openstack%2Fcinder-specs~master~Ib9058b89fefaad100a8f98e606bf7d02f2d781da","project":"openstack/cinder-specs","branch":"master","topic":"bp/sizing-encrypted-volumes","attention_set":{"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-10-26 14:40:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9535\u003e replied on the change","reason_account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib9058b89fefaad100a8f98e606bf7d02f2d781da","subject":"Sizing encrypted volumes","status":"MERGED","created":"2020-09-04 15:50:09.000000000","updated":"2023-10-26 14:40:28.000000000","submitted":"2020-12-18 20:40:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"749991-bp/sizing-encrypted-volumes","meta_rev_id":"b7e760c9744c93b4ba327d3080ca4b72ef95a0b7","_number":749991,"virtual_id_number":749991,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 20:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-12-17 16:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2020-12-18 20:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2020-12-18 20:26:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 05:50:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-28 15:33:34.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-10-28 15:34:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-04 19:36:46.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-12-10 21:16:27.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-12-18 19:55:37.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"43db46df6378aeef82657f5d70e9eca3d93fbe87","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-09-04 15:50:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15292d69a34a7baacc6d7d75079a566a44a241ab","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-09-04 15:52:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"af1adcc927d4515970941dff0da070d6eac1a61f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:00:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fcf0436874459bb6e4df3d0b7f1144 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f5dcc1d2c941a7b04c5f5bcef51647 : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"bf8336fe8bb71086db8e038631b709e3e14841a6","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-10-28 05:41:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6a3edcd2c7162a322de46532b62d1b875ce0e34d","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-10-28 05:43:08.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"136b6fd3e66e4e395ec25289cd8c39642cc66518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 05:50:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cda93518004dd7bb4664f60aaeca09 : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b732b1e9e8643cb914de3a4808203bc : FAILURE in 3m 11s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/wallaby/sizing-encrypted-volumes.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/wallaby/sizing-encrypted-volumes.rst","accounts_in_message":[],"_revision_number":4},{"id":"91cf203ed58d3de45d1793bf5d915fd958e98519","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-10-28 12:57:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c6cdb6d9b240eae6f5f17d2c7933c96cfa18a96a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 13:02:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8135e50b95df4008b99dddb44f00ecb0 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d8aced5b76440ca4f89394cf66f7d0 : FAILURE in 3m 36s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/wallaby/sizing-encrypted-volumes.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/wallaby/sizing-encrypted-volumes.rst","accounts_in_message":[],"_revision_number":4},{"id":"8f7e9e9957a487dc918aec5caf91ba5cacd35c6f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-28 15:34:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9dc1486a01293e836c3cd73f9af1a7a44a26f8e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-10-28 15:37:58.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"08fc2b53736009adc27e31f2ac9caf4c772422c2","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-11-04 19:36:46.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"620f1fcea74371c1262bb1bfc992ee9406befd80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-11-25 02:24:15.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"22e6950dfc1e3ce22aa6aae5fc2f2afed5c50372","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-11-25 02:26:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d767c22ab9a9507743cffadbbf1d029841b74053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 02:28:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977d012cbf404911bca28df5318d5b63 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/422b796209214230bc5f581016ce11a7 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8117c20a0c3c34fa472be707ebdff860bbded97f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-10 21:16:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(6 comments)\n\nI think this is a good proposal.  More detailed comments are inline.\n\nIt would be nice if we could increase the size just the right amount, but I think you are correct that for reasons of uniformity across drivers and for accounting purposes, we need to stick with the 1 GB increments we\u0027re already using.","accounts_in_message":[],"_revision_number":5},{"id":"c4d590343a408236a0f65ec8d8e06c265e7e1379","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-14 21:26:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6fb6543ddbbe14165fa15b4f441a8dc1c5745427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-15 05:01:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c71ca2033aa179574382b61835ff08731c3d6b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-15 05:07:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8aa5b4ebec43c0fb949c53cff2d1d8f18d03baef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 05:11:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de15414bcc0444090a343504a7c0d8e : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37dd88f641f247ca97f2e8306119f80d : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":7},{"id":"430fd87f5703eff0f5a5da297baf8ba868ad2c10","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-15 14:22:37.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nRequest for a clarification inline; otherwise, I think this looks good.","accounts_in_message":[],"_revision_number":7},{"id":"b2dd8b609df01f0a9fa5dd63d0baac77c81c9cf6","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-15 16:11:12.000000000","message":"Patch Set 7: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3b7f5f4ede72b4f0ac8042c525a7665ed925b5ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-15 16:33:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1ebb99aef8cb3b1a87696515c8ff7d062fe1dda0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 16:40:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e315549afcb4b798eadd480c3214438 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce5a56a8794415da90cacdd55462223 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":8},{"id":"22ed76f6e96a1ae2421fc193a3041c9606ced7d2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-15 16:59:30.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)\n\nCaught a couple of minor things, otherwise I think this is ready to go.","accounts_in_message":[],"_revision_number":8},{"id":"8cdecc6e9dc61853a076b15e6cde1f74a57a36d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-15 17:07:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"733a90c7890382cd0f76d015979e8b8f880c8a24","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-15 17:17:00.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)\n\nOne last thing ... promise!","accounts_in_message":[],"_revision_number":9},{"id":"19939466dfbc6091df31cba7054d1c6cea7e29c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-15 17:24:04.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5aa26cb3b7765a1cdda83768135cb2ab9d738c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 17:49:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8bab3f46034430ad52807c33f4da41 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f957e5eb314a1e9af345363f632ce5 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":10},{"id":"ab26932b516cf0992bcff40cff8763a020400ba7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-15 20:07:31.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)\n\nNit noted inline.  Content is great!","accounts_in_message":[],"_revision_number":10},{"id":"1f28e6a894e069d5c926dc7a0df3a70097403fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-15 22:02:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bb1a74144dff8b59ec08c32d7116365e1454bdad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 22:17:37.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d823e7c69da148a1aac336cc46c65e90 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0dee401434f44c5a386a9bdc66d1615 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":11},{"id":"c3901534b5d0683dfa0e870dab307aa055ce6318","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-16 19:41:04.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)\n\nCorrection noted inline (you can do it in a follow-up).","accounts_in_message":[],"_revision_number":11},{"id":"a5fd4c6b3b98aee53a2175980744f93ac009cb83","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-17 15:31:44.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7749b5b30da9e69d6554536dda14ff6e873664d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-12-17 15:31:48.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"daf397ae65de1056d1fa40e32a6488701abd816f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:40:18.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c0139392a14095bcfe6f272ca8d002 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d983ff400640739996e95e89e46c9a : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":12},{"id":"3c269d63fff188f34b6ee8f7d17f3e70768ca97b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-17 16:51:16.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM.  Thanks, Sofia!","accounts_in_message":[],"_revision_number":12},{"id":"98e8205072498ff0c6a345e04c21cd2930b3eaea","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-12-18 20:26:18.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"8244a14f982a0edfac4ed7b72fc4aba75e64a46d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 20:26:39.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"ce4ed5bc7fd9d188e81f1fc4c23013a941048ed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 20:40:31.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63afe5f374b64391b2779a7adcbdab65 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814a981647eb465db493d5b7231d165c : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":12},{"id":"0ab086b8dfa22c0f2bbb9e4d3feab997f7dfc1ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 20:40:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"fdcf2a3380aaac31b2094af9ec87d6b25fc8f803","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 20:42:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e5a070b4da214050a142d7baf8c1100a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":12},{"id":"b7e760c9744c93b4ba327d3080ca4b72ef95a0b7","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-10-26 14:40:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"bcdb3efdcf0f0d66f93ef0c64c247c21033ea213","revisions":{"3f914496286309262e6e97a4f1cd98e50663c3f8":{"kind":"REWORK","_number":1,"created":"2020-09-04 15:50:09.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/1"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:50:01.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3f914496286309262e6e97a4f1cd98e50663c3f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3f914496286309262e6e97a4f1cd98e50663c3f8"}]},"branch":"refs/heads/master"},"11b2b559e53b8d606fb65fe633fdb3128f827e5e":{"kind":"REWORK","_number":2,"created":"2020-09-04 15:52:22.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/2"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:52:10.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/11b2b559e53b8d606fb65fe633fdb3128f827e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/11b2b559e53b8d606fb65fe633fdb3128f827e5e"}]},"branch":"refs/heads/master"},"9ca36a6e2025e6e8a28572b901031872bc05ba02":{"kind":"REWORK","_number":3,"created":"2020-10-28 05:41:40.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/3"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-10-28 05:41:22.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9ca36a6e2025e6e8a28572b901031872bc05ba02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9ca36a6e2025e6e8a28572b901031872bc05ba02"}]},"branch":"refs/heads/master"},"7dbdf2384258d93e57351b74f0dd00c39370a6d6":{"kind":"REWORK","_number":4,"created":"2020-10-28 05:43:08.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/4"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-10-28 05:43:04.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7dbdf2384258d93e57351b74f0dd00c39370a6d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7dbdf2384258d93e57351b74f0dd00c39370a6d6"}]},"branch":"refs/heads/master"},"bf4870b896acab85214c3d7e08a0707fd61d39f7":{"kind":"REWORK","_number":5,"created":"2020-11-25 02:24:15.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/5"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-11-25 02:23:59.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bf4870b896acab85214c3d7e08a0707fd61d39f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bf4870b896acab85214c3d7e08a0707fd61d39f7"}]},"branch":"refs/heads/master"},"eb63223d97e555fcafbd713ab91696893a238834":{"kind":"REWORK","_number":6,"created":"2020-12-15 05:01:01.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/6"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-12-15 05:00:55.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/eb63223d97e555fcafbd713ab91696893a238834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/eb63223d97e555fcafbd713ab91696893a238834"}]},"branch":"refs/heads/master"},"f59e2f5d4669b96e7c2c5f1af3475ea66e3def53":{"kind":"REWORK","_number":7,"created":"2020-12-15 05:07:02.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/7"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-12-15 05:06:39.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f59e2f5d4669b96e7c2c5f1af3475ea66e3def53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f59e2f5d4669b96e7c2c5f1af3475ea66e3def53"}]},"branch":"refs/heads/master"},"230aafbd2e048b1c0fc2382be9eff43973df8eeb":{"kind":"REWORK","_number":8,"created":"2020-12-15 16:33:31.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/8"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-12-15 16:33:08.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/230aafbd2e048b1c0fc2382be9eff43973df8eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/230aafbd2e048b1c0fc2382be9eff43973df8eeb"}]},"branch":"refs/heads/master"},"88c1b3d27f519a49c2fd88fbf695a9bd5829a2b9":{"kind":"REWORK","_number":9,"created":"2020-12-15 17:07:13.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/9"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-12-15 17:06:53.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/88c1b3d27f519a49c2fd88fbf695a9bd5829a2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/88c1b3d27f519a49c2fd88fbf695a9bd5829a2b9"}]},"branch":"refs/heads/master"},"d598bb14e84c51f95ff74c0203208f6b6c42aa5d":{"kind":"REWORK","_number":10,"created":"2020-12-15 17:24:04.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/10"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-12-15 17:23:37.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d598bb14e84c51f95ff74c0203208f6b6c42aa5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d598bb14e84c51f95ff74c0203208f6b6c42aa5d"}]},"branch":"refs/heads/master"},"e2b971c1f14fd8b1d137417d263f8e249ba5bd7c":{"kind":"REWORK","_number":11,"created":"2020-12-15 22:02:46.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/11"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-12-15 22:02:24.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2b971c1f14fd8b1d137417d263f8e249ba5bd7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2b971c1f14fd8b1d137417d263f8e249ba5bd7c"}]},"branch":"refs/heads/master"},"bcdb3efdcf0f0d66f93ef0c64c247c21033ea213":{"kind":"REWORK","_number":12,"created":"2020-12-17 15:31:48.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/91/749991/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/91/749991/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/91/749991/12"}}},"commit":{"parents":[{"commit":"c368eb990d31cf372acee8b78db2662f6de237a4","subject":"Merge \"Add wallaby directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c368eb990d31cf372acee8b78db2662f6de237a4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-04 15:47:42.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-12-17 15:31:33.000000000","tz":0},"subject":"Sizing encrypted volumes","message":"Sizing encrypted volumes\n\nThis feature will support retype/migrate unencrypted volumes to\nencrypted type.\n\nChange-Id: Ib9058b89fefaad100a8f98e606bf7d02f2d781da\nImplements: blueprint sizing-encrypted-volumes\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bcdb3efdcf0f0d66f93ef0c64c247c21033ea213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bcdb3efdcf0f0d66f93ef0c64c247c21033ea213"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
