)]}'
{"id":"openstack%2Fcinder-specs~758375","triplet_id":"openstack%2Fcinder-specs~master~Ib8db275178b258140cc628aa0f940368982e86f1","project":"openstack/cinder-specs","branch":"master","topic":"bp/volume-safe-delete","hashtags":[],"change_id":"Ib8db275178b258140cc628aa0f940368982e86f1","subject":"Support volume safe delete","status":"ABANDONED","created":"2020-10-15 09:26:36.000000000","updated":"2020-11-25 08:35:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76e1f760b8467cc13ce74b1887637e1fa6079bf1","_number":758375,"virtual_id_number":758375,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-10-15 11:20:32.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-15 09:40:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":-1,"date":"2020-10-15 09:43:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2020-11-05 10:59:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 09:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-15 09:43:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-05 10:59:33.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"987eb675f7f30c7bc317378747eaec8a23894f2e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-15 09:26:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0548e2e6a2107dbc779bd14950e2d9f8a4f0dba4","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-15 09:32:07.000000000","message":"Topic set to bp/volume-safe-delete","accounts_in_message":[],"_revision_number":1},{"id":"b65f6344cb5342fafc4f2923e1ba1549ab843198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 09:40:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db8c71b0b114d418ac23d5c4a1e4830 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e50fe9db324154b5e6d18c029e4565 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3e814d93947669b1290f6bd0adbf6e773742cbd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-15 09:43:21.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is actually a security bug for any backend driver that does not automatically take care of this. This should not be an extra step someone needs to know to take.","accounts_in_message":[],"_revision_number":1},{"id":"c51aa4049c865893d695aaa2487ebe86b10fde52","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-15 11:20:32.000000000","message":"Patch Set 1:\n\nhi, Sean McGinnis, thanks for your review.\njust because any backend driver that does not automatically take care of this, so we added this feature to the cloud platform side, allowing users to implement it through the cloud platform without having to rely on back-end storage.\ni think as an end user, he need this, if the data in the volume is very sensitive.","accounts_in_message":[],"_revision_number":1},{"id":"2865a4659b6afa08502df35b2591972548beffca","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-11-05 10:59:33.000000000","message":"Patch Set 1: Code-Review-2\n\nThis was discussed during the Wallaby PTG [1] and our documentation is being improved to explain the security aspects of this feature.\n\nIf you feel that this should still be addressed or have questions/concerns please add it to the next IRC meeting agenda [3] and come to the meeting to let us know.\n\nInfo on the Cinder IRC meetings can be found at https://wiki.openstack.org/wiki/CinderMeetings\n\n[1]: https://wiki.openstack.org/wiki/CinderWallabyPTGSummary#Two_proposed_specs_on_the_same_topic_.28mutually_assured_destruction.29\n[2]: https://review.opendev.org/761535 \n[3]: https://etherpad.opendev.org/p/cinder-wallaby-meetings","accounts_in_message":[],"_revision_number":1},{"id":"76e1f760b8467cc13ce74b1887637e1fa6079bf1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-11-25 08:35:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5fa2629f89125eed3719ade19be44e7758e4b681","revisions":{"5fa2629f89125eed3719ade19be44e7758e4b681":{"kind":"REWORK","_number":1,"created":"2020-10-15 09:26:36.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/758375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/75/758375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/75/758375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/75/758375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/75/758375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/75/758375/1"}}},"commit":{"parents":[{"commit":"dac5f63df92936d29c7c1cc35c18a53bb56d80a3","subject":"Update: default type override spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/dac5f63df92936d29c7c1cc35c18a53bb56d80a3"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-15 09:24:43.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-15 09:24:43.000000000","tz":480},"subject":"Support volume safe delete","message":"Support volume safe delete\n\nChange-Id: Ib8db275178b258140cc628aa0f940368982e86f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5fa2629f89125eed3719ade19be44e7758e4b681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5fa2629f89125eed3719ade19be44e7758e4b681"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
