)]}'
{"id":"openstack%2Fcinder-specs~764628","triplet_id":"openstack%2Fcinder-specs~master~Ica9bf5fd497f64ddf38d65805857b88e196bea29","project":"openstack/cinder-specs","branch":"master","topic":"bp/volume-upload-to-glance-by-clone","hashtags":[],"change_id":"Ica9bf5fd497f64ddf38d65805857b88e196bea29","subject":"When using ceph, the volume is uploaded to glance by cloning","status":"ABANDONED","created":"2020-11-30 10:34:26.000000000","updated":"2020-12-17 10:10:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f960c5c0c69e032f0492786f0d4fc4b4a645090","_number":764628,"virtual_id_number":764628,"owner":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-02 03:13:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 10:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 15:10:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"4eb9ec6881d654be3eedcc1ce635e26071742cf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"date":"2020-11-30 10:34:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06ed235bae79e644c1295a0664f57a07b2c05d5","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"date":"2020-11-30 10:43:37.000000000","message":"Assignee added: chengsheng \u003cchengsheng@chinatelecom.cn\u003e","accounts_in_message":[],"_revision_number":1},{"id":"4205fc1b5714174bd15f7bb100308c58736c513a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 10:44:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda4ef6b5fda43eb8fba2a3214e56ef1 : FAILURE in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2dd24a2cb2243d3a51601632af07f48 : FAILURE in 3m 39s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/wallaby/volume-upload-to-glance-by-clone.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/wallaby/volume-upload-to-glance-by-clone.rst","accounts_in_message":[],"_revision_number":1},{"id":"c93d17c5140954a78f552ff7a844398d0502524a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"date":"2020-12-01 02:07:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"210334d9029573a0e4eac97413c770b55091f95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 02:15:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8ebf0496884725b7f71a6bea7ea000 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf1568e2e6642e89b9fc6f2e4ff7382 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ca3b2d9eda931a37801c15888395e82de7526618","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"date":"2020-12-02 03:02:31.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8d2e308f53627d0f38969bd96b21f747d8cfd10f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 03:13:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf27ac961c940cca51791ec2f6a4185 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f0416e6a4ea401aac4052f441d9600b : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"1b8520a51167ec3772e48613db3b38935b78b3b2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-16 15:10:52.000000000","message":"Patch Set 3:\n\nAs a driver-specific optimization, this doesn\u0027t need a full spec.  You can just track it on the Launchpad blueprint:\n\nhttps://blueprints.launchpad.net/cinder/+spec/volume-upload-to-glance-by-clone\n\nand abandon this review.","accounts_in_message":[],"_revision_number":3},{"id":"7f960c5c0c69e032f0492786f0d4fc4b4a645090","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"date":"2020-12-17 10:10:49.000000000","message":"Abandoned\n\nAs a driver-specific optimization, this doesn\u0027t need a full spec","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b298c1db88f20a12da2abb2a20dd8e65a989fd9b","revisions":{"50fc15736b7bac19e5c97a65bf049135548fa639":{"kind":"REWORK","_number":1,"created":"2020-11-30 10:34:26.000000000","uploader":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"ref":"refs/changes/28/764628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/28/764628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/1"}}},"commit":{"parents":[{"commit":"873135ff0529e57a1177f00def79b01c167ce740","subject":"Merge \"Retarget specs for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/873135ff0529e57a1177f00def79b01c167ce740"}]}],"author":{"name":"Cheng Sheng","email":"chengsheng@chinatelecom.cn","date":"2020-11-30 10:24:04.000000000","tz":480},"committer":{"name":"Cheng Sheng","email":"chengsheng@chinatelecom.cn","date":"2020-11-30 10:33:50.000000000","tz":480},"subject":"When using ceph, the volume is uploaded to glance by cloning","message":"When using ceph, the volume is uploaded to glance by cloning\n\nImplements: blueprint volume-upload-to-glance-by-clone\nChange-Id: Ica9bf5fd497f64ddf38d65805857b88e196bea29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/50fc15736b7bac19e5c97a65bf049135548fa639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/50fc15736b7bac19e5c97a65bf049135548fa639"}]},"branch":"refs/heads/master"},"d2e087405359efe5355f216e9657873fadfa03a4":{"kind":"REWORK","_number":2,"created":"2020-12-01 02:07:32.000000000","uploader":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"ref":"refs/changes/28/764628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/28/764628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/2"}}},"commit":{"parents":[{"commit":"873135ff0529e57a1177f00def79b01c167ce740","subject":"Merge \"Retarget specs for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/873135ff0529e57a1177f00def79b01c167ce740"}]}],"author":{"name":"Cheng Sheng","email":"chengsheng@chinatelecom.cn","date":"2020-11-30 10:24:04.000000000","tz":480},"committer":{"name":"Cheng Sheng","email":"chengsheng@chinatelecom.cn","date":"2020-12-01 02:06:54.000000000","tz":480},"subject":"When using ceph, the volume is uploaded to glance by cloning","message":"When using ceph, the volume is uploaded to glance by cloning\n\nImplements: blueprint volume-upload-to-glance-by-clone\nChange-Id: Ica9bf5fd497f64ddf38d65805857b88e196bea29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d2e087405359efe5355f216e9657873fadfa03a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d2e087405359efe5355f216e9657873fadfa03a4"}]},"branch":"refs/heads/master"},"b298c1db88f20a12da2abb2a20dd8e65a989fd9b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-02 03:02:31.000000000","uploader":{"_account_id":12203,"name":"chengsheng","email":"chengsheng@chinatelecom.cn","username":"chengsheng"},"ref":"refs/changes/28/764628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/28/764628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/28/764628/3"}}},"commit":{"parents":[{"commit":"873135ff0529e57a1177f00def79b01c167ce740","subject":"Merge \"Retarget specs for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/873135ff0529e57a1177f00def79b01c167ce740"}]}],"author":{"name":"Cheng Sheng","email":"chengsheng@chinatelecom.cn","date":"2020-11-30 10:24:04.000000000","tz":480},"committer":{"name":"Cheng Sheng","email":"chengsheng@chinatelecom.cn","date":"2020-12-02 03:00:38.000000000","tz":480},"subject":"When using ceph, the volume is uploaded to glance by cloning","message":"When using ceph, the volume is uploaded to glance by cloning\n\nWhen using ceph, to create a image from a volume,\nthe volume will be downloaded to a local temporary\ndirectory, and then uploaded to glance after conversion.\nHowever, if you create a raw format image, you only\nneed to upload it to glance through clone.\n\nImplements: blueprint volume-upload-to-glance-by-clone\nChange-Id: Ica9bf5fd497f64ddf38d65805857b88e196bea29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b298c1db88f20a12da2abb2a20dd8e65a989fd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b298c1db88f20a12da2abb2a20dd8e65a989fd9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
