)]}'
{"id":"openstack%2Fcinder-specs~766716","triplet_id":"openstack%2Fcinder-specs~master~I7a32f05499406d5911e60d219ffb254b24b7d929","project":"openstack/cinder-specs","branch":"master","attention_set":{"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2022-11-02 06:57:43.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7a32f05499406d5911e60d219ffb254b24b7d929","subject":"Reset state robustification","status":"MERGED","created":"2020-12-11 11:06:19.000000000","updated":"2022-11-02 06:57:43.000000000","submitted":"2020-12-18 14:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"766716","meta_rev_id":"d15b33b56427d2efbe1b071d7c8a18a819ae1ceb","_number":766716,"virtual_id_number":766716,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 14:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2020-12-16 20:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-12-17 09:17:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-18 14:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-18 14:07:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-11 11:16:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 16:07:36.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-12-17 09:17:27.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"CC"},{"updated":"2020-12-17 09:17:37.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"REVIEWER"},{"updated":"2020-12-18 14:06:18.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2020-12-18 14:07:23.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-11-02 06:57:43.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"}],"messages":[{"id":"a8b73bb11322b18688afc2ab154ff492bedffa84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2020-12-11 11:06:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b490fea6b6c677fbf322a82a1e053c29166068a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 11:16:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af97663b4c3b4e6f9e7bce02897bdfbc : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71fcf41789248ff8fb64e86666c4b83 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"768b4fa715357d7e893c8fe52f39da640e770e19","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-11 16:07:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nA few nits noted inline.  I think this is a good proposal, and even though it\u0027s only currently focused on a small number of cases, it will give cinder a framework that can be extended to handle other situations as they come to our attention.  So I\u0027m in favor of this being worked on in Wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"5816947074828dbfc078ca88131aaab031503ac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2020-12-14 06:57:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8340e8dff5c57f4dadaae0c71d93cb8a8e25df6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 07:12:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0855d79c3847118b389c3de04a60ba : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b0ff3570a94e8cba7d02fc027bab17 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b1a33e7a910906017cc832f512c034474d2095c0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-14 21:28:53.000000000","message":"Patch Set 2: Code-Review+2\n\nYou\u0027ve addressed my concerns.  Thanks!  As I said in my previous comment, I think this is a good proposal.","accounts_in_message":[],"_revision_number":2},{"id":"0be0b71a8aff61898238cfde14a011c11299d3d0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-14 21:35:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nTwo small changes requested inline.","accounts_in_message":[],"_revision_number":2},{"id":"b2a8154e8590daea60e7423ed2dc44db4cbbc7c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2020-12-15 04:42:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8d9cf4d61e4e8de76a95f2a2bb2fe17b9f0c3c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 04:51:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a615f89b59841d9a96c8424e43e3d70 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/801d3ba74b494eb793befd6fbcfab4f5 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"38c1eabd624db2957ecb22832a083fec12f1f8a9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-16 20:35:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"45b02450d22bce542321f868be5b682aeb6a87d1","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-17 09:17:27.000000000","message":"Patch Set 3:\n\nDoc is about reset the state of attached volumes available, LGTM","accounts_in_message":[],"_revision_number":3},{"id":"9fc3ae2a5dfd6b1123d148779052eabaaa47c58b","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-17 09:17:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9d2d282447bab1835838bd0935ae03c5eae12f28","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-12-18 14:06:18.000000000","message":"Patch Set 3:\n\nI think this is a good improvement.  Lets move this forward.","accounts_in_message":[],"_revision_number":3},{"id":"7c640150769b99b0201b647190248281a9e6b79b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-12-18 14:07:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nHelps when I actually vote.  :-)","accounts_in_message":[],"_revision_number":3},{"id":"0f2451ee1f4a172e4658cd8b9161ff70f3106bf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:07:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1c3907fc6d75f94a9cbf406efdf08cb4133cd233","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:13:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d786151d2adf4a5eba382631fcee02b2 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af98d0d47e747db992a7010a7f79ee3 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"aff2cb1a56a62dc93f67e712176d7881881e2309","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:13:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7462c03521e5f36d806f3697955aa44e06b867f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:15:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e1c0d65d60254bc991364d12de552151 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3},{"id":"d15b33b56427d2efbe1b071d7c8a18a819ae1ceb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-02 06:57:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7b9f133c5dc3069a06f5fb3ce24247621d6390eb","revisions":{"02f3453da364ea397fe3e8801f1763f2169a2e38":{"kind":"REWORK","_number":1,"created":"2020-12-11 11:06:19.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/16/766716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/16/766716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/1"}}},"commit":{"parents":[{"commit":"30466d896b0a72fafe5461f88ef311984f7defb8","subject":"Merge \"Add clarification about re-proposing specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/30466d896b0a72fafe5461f88ef311984f7defb8"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2020-12-11 11:04:46.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2020-12-11 11:06:14.000000000","tz":330},"subject":"Reset state robustification","message":"Reset state robustification\n\nChange-Id: I7a32f05499406d5911e60d219ffb254b24b7d929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/02f3453da364ea397fe3e8801f1763f2169a2e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/02f3453da364ea397fe3e8801f1763f2169a2e38"}]},"branch":"refs/heads/master"},"53b47932038e49c56050c57b081d93d8c89a301b":{"kind":"REWORK","_number":2,"created":"2020-12-14 06:57:07.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/16/766716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/16/766716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/2"}}},"commit":{"parents":[{"commit":"30466d896b0a72fafe5461f88ef311984f7defb8","subject":"Merge \"Add clarification about re-proposing specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/30466d896b0a72fafe5461f88ef311984f7defb8"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2020-12-11 11:04:46.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2020-12-14 06:56:40.000000000","tz":330},"subject":"Reset state robustification","message":"Reset state robustification\n\nChange-Id: I7a32f05499406d5911e60d219ffb254b24b7d929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/53b47932038e49c56050c57b081d93d8c89a301b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/53b47932038e49c56050c57b081d93d8c89a301b"}]},"branch":"refs/heads/master"},"7b9f133c5dc3069a06f5fb3ce24247621d6390eb":{"kind":"REWORK","_number":3,"created":"2020-12-15 04:42:26.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/16/766716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/16/766716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/16/766716/3"}}},"commit":{"parents":[{"commit":"30466d896b0a72fafe5461f88ef311984f7defb8","subject":"Merge \"Add clarification about re-proposing specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/30466d896b0a72fafe5461f88ef311984f7defb8"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2020-12-11 11:04:46.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2020-12-15 04:42:05.000000000","tz":330},"subject":"Reset state robustification","message":"Reset state robustification\n\nCo-authored-by: Eric Harney \u003ceharney@redhat.com\u003e\n\nChange-Id: I7a32f05499406d5911e60d219ffb254b24b7d929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7b9f133c5dc3069a06f5fb3ce24247621d6390eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7b9f133c5dc3069a06f5fb3ce24247621d6390eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
