)]}'
{"id":"openstack%2Fcinder-specs~778436","triplet_id":"openstack%2Fcinder-specs~master~I846fb0432232feb4f9ea5d07e8fa86bb4e622372","project":"openstack/cinder-specs","branch":"master","hashtags":[],"change_id":"I846fb0432232feb4f9ea5d07e8fa86bb4e622372","subject":"Add xena directory for specs","status":"MERGED","created":"2021-03-03 11:41:55.000000000","updated":"2021-03-04 01:52:31.000000000","submitted":"2021-03-04 01:51:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"778436","meta_rev_id":"85170a395021b7f5fd7df625f6206a4ace6a6210","_number":778436,"virtual_id_number":778436,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 01:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-04 01:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-04 01:37:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 11:57:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 23:35:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"d633dd78f2854a51e3aee41667c00c1b47093413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-03 11:41:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"761b6f9bd81a46feb46695484b0cd174ce8121ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 11:57:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50aff411ffb4c668bca95ed8f44d436 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654194c8899949a2a6f018181e242bef : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4a8b26a1152b341955ce88b896a1a601de8bc626","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-03 23:35:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nA few things noted inline.  Thanks for posting this, I should have done it a month ago!","accounts_in_message":[],"_revision_number":1},{"id":"d71beb3d84428f52df8ff6440f9bf548982b6600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-04 01:10:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6af8fde80c75bf7960c655819dcf56ff86420eb8","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-04 01:10:46.000000000","message":"Patch Set 2:\n\nOk, got it, thanks very much","accounts_in_message":[],"_revision_number":2},{"id":"173d84de62b2cea2d9e4a2dcb983ec7f95a82cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:26:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cea3f800f1342908dcbd6de99a3754b : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb80b35277a47d896df32a499fb92bc : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1cdbd4e7a0a99a423867b605089a231ecc868673","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 01:37:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nPurely a docs change, so I\u0027m going to ninja-approve this one.  Thanks, Xuan.","accounts_in_message":[],"_revision_number":2},{"id":"cb3a084f387a1d68b64449c9d4efef7449e503a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:40:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"535150aeafea1feb09b15320839ea60885681af6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:51:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e36b8e62804b919888912f90737d27 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9f85d1b78145059c72354717e21094 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1de32e37a5db3c92f91b50dc4f5fd3528d3caeda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:51:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"85170a395021b7f5fd7df625f6206a4ace6a6210","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:52:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/8c51ed7fa36a435ea0e8efce72160d90 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2e1f0761bc41104e6a156cd921424708613b877","revisions":{"3b8d1b73e464b271e391938126f29d22c04bb736":{"kind":"REWORK","_number":1,"created":"2021-03-03 11:41:55.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/36/778436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/36/778436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/1"}}},"commit":{"parents":[{"commit":"bb8e6bc242c1ca90f9cce6f40e18d6e439720223","subject":"Merge \"NVMe monitoring and healing agent for NVMe connector.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bb8e6bc242c1ca90f9cce6f40e18d6e439720223"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-03-03 11:41:43.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-03-03 11:41:43.000000000","tz":480},"subject":"Add xena directory for specs","message":"Add xena directory for specs\n\nChange-Id: I846fb0432232feb4f9ea5d07e8fa86bb4e622372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b8d1b73e464b271e391938126f29d22c04bb736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b8d1b73e464b271e391938126f29d22c04bb736"}]},"branch":"refs/heads/master"},"e2e1f0761bc41104e6a156cd921424708613b877":{"kind":"REWORK","_number":2,"created":"2021-03-04 01:10:28.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/36/778436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/36/778436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/36/778436/2"}}},"commit":{"parents":[{"commit":"bb8e6bc242c1ca90f9cce6f40e18d6e439720223","subject":"Merge \"NVMe monitoring and healing agent for NVMe connector.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/bb8e6bc242c1ca90f9cce6f40e18d6e439720223"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-03-03 11:41:43.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-03-04 01:09:43.000000000","tz":480},"subject":"Add xena directory for specs","message":"Add xena directory for specs\n\nChange-Id: I846fb0432232feb4f9ea5d07e8fa86bb4e622372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2e1f0761bc41104e6a156cd921424708613b877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2e1f0761bc41104e6a156cd921424708613b877"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
