)]}'
{"id":"openstack%2Fcinder-specs~796166","triplet_id":"openstack%2Fcinder-specs~master~Id236326dca9ee08aad8427ea6cb400e25e7ac09e","project":"openstack/cinder-specs","branch":"master","hashtags":[],"change_id":"Id236326dca9ee08aad8427ea6cb400e25e7ac09e","subject":"Expose ``user visible`` extra specs","status":"MERGED","created":"2021-06-13 20:21:04.000000000","updated":"2021-06-30 13:09:03.000000000","submitted":"2021-06-30 13:07:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":113,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"796166","meta_rev_id":"5bf5c090e608c4f08785130671e5b9c9775fcad2","_number":796166,"virtual_id_number":796166,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 13:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2021-06-24 15:16:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2021-06-24 14:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-24 20:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-06-24 14:47:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2021-06-25 03:30:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-13 20:22:26.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"REVIEWER"},{"updated":"2021-06-13 20:22:26.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-06-13 20:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 13:45:39.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-06-15 13:54:16.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-06-16 01:48:24.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-06-16 22:31:21.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-06-17 19:54:15.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2021-06-23 15:20:11.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"d903d0274ea593eed48c8a4aab919a4f47e06b59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-13 20:21:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41e5ae4d6207c05a9d10e02235a72af4f443b9ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 20:28:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33688be8ca6a4989804098242cdaf71c : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f994b3e45f64d91a01f4cac2eb393eb : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"04b8727d5b7bff7b49fecc5deb57aaf2eb732889","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-14 14:39:38.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8aff474fa00f99de2fcbb5f165bbb1257f071b7e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-15 11:24:45.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nThanks Alan! I agree with your review comments.","accounts_in_message":[],"_revision_number":1},{"id":"f06f12131f9ee414aa869d1bd10a73be4c321238","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-16 01:48:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef65f4639aa2b6e5357f8d7da4e0001fbbbb9cda","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-16 02:17:54.000000000","message":"Patch Set 1:\n\n(8 comments)\n\nThanks for the review Rajesh, especially for the careful attention to the extra spec names.  I\u0027m not sure about using \u0027project visible\u0027 rather than \u0027tenant visible\u0027, especially in our new world with Domain and System columns alongside Project, but I\u0027m not dug in about it.  I chose \u0027tenant\u0027 as the traditional term for the concept of non-administrative users of the cloud.  But I\u0027m open to using whatever terminology the Cinder community thinks best.\n\nAssuming we can get that straightened out, I think the main task will be (as abishop noted) to enumerate a clear set of tenant/project visible extra specs that will be exposed to non administrative users with the new microversion.  These should be abstract, backend independent, and indicate features or capabilities that a regular user needs to know in order for their volume creates not to be successful.  If you can help us identify these (I\u0027m a bit stale on Cinder), that would be very helpful.","accounts_in_message":[],"_revision_number":1},{"id":"97c1cc4d21909393aca0566f9e6b63f40f1797d5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-16 12:12:15.000000000","message":"Patch Set 1:\n\nSince in the old days \u0027tenant\u0027 was used in keystone for what is now called \u0027project\u0027, we\u0027ll need to guard against the interpretation that what we really mean here are project-scoped extra specs.  What we intend instead are extra specs that are (also) visible to non-admins.  This is a matter of keystone role, rather than of keystone scope.  The idea is that there is a subset of abstract, backend-independent extra specs that should be visible to all the keystone roles, to members and readers as well as administrators.  We have been putting the point by saying they would be visible to tenants, the users who rent or are granted use of the infrastructure but who do not own the infrastructure or administer it on behalf of the owners.","accounts_in_message":[],"_revision_number":1},{"id":"874cb4596529bebde41b0355a7df5e983802b88f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-16 12:49:09.000000000","message":"Patch Set 1:\n\nI see both sides:\n\n1. Tenant visible is meant to convey the specs are visible to everyone, and not just admins\n2. In its traditional context the term \"tenant\" is deprecated in favor of \"project\"\n\n\"Project visible\" satisfies 2, but at the expense of 1.\n\nWhat do people think of using the term \"member visible?\" I feel is satisfies 1 without running afoul of 2.","accounts_in_message":[],"_revision_number":1},{"id":"36d5545a9ea9b737d786e2f86c311dffd52cada9","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-16 14:21:36.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I see both sides:\n\u003e \n\u003e 1. Tenant visible is meant to convey the specs are visible to everyone, and not just admins\n\u003e 2. In its traditional context the term \"tenant\" is deprecated in favor of \"project\"\n\u003e \n\u003e \"Project visible\" satisfies 2, but at the expense of 1.\n\u003e \n\u003e What do people think of using the term \"member visible?\" I feel is satisfies 1 without running afoul of 2.\n\nI\u0027m OK with member_visible but note that in the prototype patch the extra specs are visible to all roles (including reader) and that\u0027s the way I think it should be.","accounts_in_message":[],"_revision_number":1},{"id":"f163851d9fd8828d8c6f4bef2565914d5c36b3b6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-17 12:02:44.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nNew revision on the way, thanks for the helpful comments!","accounts_in_message":[],"_revision_number":1},{"id":"19a5eed62f2b5f760a6545192293681309fbe207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-17 12:11:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e7563e159c610664b7768b3b336356faaac32cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:17:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5b551902154faa8a7849bcf8491f90 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eee1e40800c47148af649cfb0da7913 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"250f9b1b4af59100ca30c8841c8ab147e83188f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-17 19:43:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2243346c3ee304c1758bb751db77a1b6e6f4cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:51:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d27e1bbe9084de4985827040cd7d082 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fc5c2a4eba42e29a4fe448518856e0 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"719b7e0ab7894e2a1cb487a71438a3a66323b401","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-17 19:54:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ab472f4226026d49faaa543d1c2152793eff4cec","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-18 13:50:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nBasically looks good, but there are a few questions I\u0027d like you to address.","accounts_in_message":[],"_revision_number":3},{"id":"74e0ada54aebcb5c7d631625f0a7d68c1d857364","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-18 14:17:18.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f0b471cb8969a9855a4f39ba2c65dfc80d27f126","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2021-06-18 15:51:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nThis looks good to me. I would be able to implement ReadWriteMany support in OpenShift for cinder backends based on this.","accounts_in_message":[],"_revision_number":3},{"id":"cf77ff63b96d42937767633a0ca415b759f73b96","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-19 19:48:35.000000000","message":"Patch Set 3:\n\n(5 comments)\n\nA few followup remarks inline.","accounts_in_message":[],"_revision_number":3},{"id":"8a4e4d19599d12cc57e07fb591cf1f0a2ca88ab4","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-21 14:29:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cc9206e288cf2895f8ada610c555efa57633380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-21 18:21:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"019a524a03fbd0c5a5e9d30cd5e9d672082dfecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 18:33:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca1ce7c45ff4b51bf15b7839afeeef1 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3cb4029426b46d086ddc850abc725e1 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"82efb47fbe4e4c4b93838ff1482fa0f3496ce7cc","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-21 19:07:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3cb84e8445a5729fdc06169f2af40f6fcb451e76","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-21 19:36:27.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2d4bdc46c2518f1c245bc86b9eddf82a33ebc724","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-22 05:02:54.000000000","message":"Patch Set 4:\n\nThe tl;dr for this is: I don\u0027t think we need an mv bump for this spec, but we do need a new policy.\n\nThese are the current policies we have that are relevant to extra-specs.\n\n(1) volume_extension:access_types_extra_specs\n\"List or show volume type with extra specs attribute\"\nAdds extra_specs to the following responses:\n    GET  /types/{type_id}\n    GET  /types\n\n(The ability to make the GET /types/* API calls is governed by other policies.)\n\n(2) volume_extension:types_extra_specs:index\n\"List type extra specs\"\n    GET  /types/{type_id}/extra_specs\n\n(3) volume_extension:types_extra_specs:show\n\"Show one specified type extra specs\"\n    GET  /types/{type_id}/extra_specs/{extra_spec_key}\n\n(4) volume_extension:types_extra_specs:create\n\"Create type extra specs\"\n    POST  /types/{type_id}/extra_specs\n\n(5) volume_extension:types_extra_specs:update\n\"Update type extra specs\"\n    PUT  /types/{type_id}/extra_specs/{extra_spec_key}\n\n(6) volume_extension:types_extra_specs:delete\n\"Delete type extra specs\"\n    DELETE  /types/{type_id}/extra_specs/{extra_spec_key}\n\n\n(I\u0027ll state here that we don\u0027t want to rely on an \"administrative context\" to determine whether content is displayed or not.  That has introduced problems in other calls, and is what the \"consistent and secure RBAC\" effort is trying to eliminate.)\n\nI think what we need to do for this spec is to add an additional policy, named something like \u0027volume_extension:types_extra_specs:read_sensitive\u0027 (\"RS\" for short)\n\nGiven \"multiattach\" as nonsensitive and \"volume_backend_name\" as sensitive:\n\n- user A who satisfies volume_extension:access_types_extra_specs (but NOT RS) will see an \"extra_specs\" element in the volume-type response containing \"multiattach\".\n- user B who satisfies volume_extension:access_types_extra_specs AND RS will see the full \"extra_specs\" element containing both \"multiattach\" and \"volume_backend_name\" in the volume-type response.\n\n\"List type extra specs\" (GET  /types/{type_id}/extra_specs) can be handled the same way, i.e., if you have permission to make the call, you see only the nonsensitive extra-specs in the response, and if you also have RS, you see the full list of extra-specs.\n\nThe default for volume_extension:types_extra_specs:read_sensitive will be admin-only.\n\nI don\u0027t think we need a microversion bump because the API requests/responses aren\u0027t being changed at all.\n\nWe probably ship with:\n\nvolume_extension:types_extra_specs:read_sensitive : admin-only\nvolume_extension:access_types_extra_specs : admin-only\nvolume_extension:types_extra_specs:index : admin-only\nvolume_extension:types_extra_specs:show : admin-only\nvolume_extension:types_extra_specs:create : admin-only\nvolume_extension:types_extra_specs:update : admin-only\nvolume_extension:types_extra_specs:delete : admin-only\n\nAnd then the behavior is the same as current behavior.  Also, that\u0027s why I think there is no point having a mv bump for this, because with the above settings, there is no difference between the pre-bump and post-bump API responses.  Whether someone sees the non-sensitive extra-specs or not depends on how the operator has configured policies, regardless of what API version is requested.\n\nOr, we can ship with:\n\nvolume_extension:types_extra_specs:read_sensitive : admin-only\nvolume_extension:access_types_extra_specs : any authorized user\nvolume_extension:types_extra_specs:index : any authorized user\nvolume_extension:types_extra_specs:show : any authorized user\nvolume_extension:types_extra_specs:create : admin-only\nvolume_extension:types_extra_specs:update : admin-only\nvolume_extension:types_extra_specs:delete : admin-only\n\nEven with the \"new\" behavior, I don\u0027t see the point of an mv bump because the new mv can be completely negated if the operator so chooses.  Also, the behavior isn\u0027t really new because the API responses haven\u0027t changed.\n\nThe tricky thing will be how to handle\n    GET  /types/{type_id}/extra_specs/{extra_spec_key}\nbut we basically do what you outlined in the spec, namely:\n(1) must satisfy volume_extension:types_extra_specs:show to make the call\n(2) if {extra_spec_key} is in the nonsensitive list, return the value\n(3) if {extra_spec_key} exists but is not in the nonsensitive list:\n    (3a) if caller satisfies volume_extension:types_extra_specs:read_sensitive then return the value\n    (3b) else, return 404\n(4) return 404\n\nThat doesn\u0027t seem too bad.\n\nI think there are no circumstances under which we want a non-admin doing create, update, or delete on extra-specs, so I don\u0027t think the RS policy has any role in those operations.","accounts_in_message":[],"_revision_number":4},{"id":"3bc9661f1976c36702870c04f2a6a232c923dfd4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-22 05:03:10.000000000","message":"Patch Set 4: Code-Review-1\n\nForgot to vote.","accounts_in_message":[],"_revision_number":4},{"id":"8280c860ad745ee9a22aeca324d6008bc9edea49","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-22 11:39:38.000000000","message":"Patch Set 4:\n\n\"We probably ship with:\n\nvolume_extension:types_extra_specs:read_sensitive : admin-only\nvolume_extension:access_types_extra_specs : admin-only\nvolume_extension:types_extra_specs:index : admin-only\nvolume_extension:types_extra_specs:show : admin-only\nvolume_extension:types_extra_specs:create : admin-only\nvolume_extension:types_extra_specs:update : admin-only\nvolume_extension:types_extra_specs:delete : admin-only\n\nAnd then the behavior is the same as current behavior.  Also, that\u0027s why I think there is no point having a mv bump for this, because with the above settings, there is no difference between the pre-bump and post-bump API responses.  Whether someone sees the non-sensitive extra-specs or not depends on how the operator has configured policies, regardless of what API version is requested.\n\nOr, we can ship with:\n\nvolume_extension:types_extra_specs:read_sensitive : admin-only\nvolume_extension:access_types_extra_specs : any authorized user\nvolume_extension:types_extra_specs:index : any authorized user\nvolume_extension:types_extra_specs:show : any authorized user\nvolume_extension:types_extra_specs:create : admin-only\nvolume_extension:types_extra_specs:update : admin-only\nvolume_extension:types_extra_specs:delete : admin-only\"\n\nI strongly advocate for shipping with a solution that changes the default behavior.  Just making a new behavior possible via policy but not changing it -- at a new release, with doc and release note even if no microversion bump -- doesn\u0027t really solve the problem because it requires cloud administrators to make an active change to do something that (in hindsight) Cinder should have been doing all along.\nMany administrators won\u0027t know to do it or bother to read that they should do it and their clouds will remain broken with respect to the use cases that drive this spec.\n\nI\u0027ll leave it to Cinder folks whether a new microversion is required when the default REST API *behavior* (expected responses) changes.  In Manila I think we would bump it even if the syntactic form of the responses isn\u0027t changed if the semantics are new.","accounts_in_message":[],"_revision_number":4},{"id":"ac2221435a7584cb21fe51563252347b14a7378a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-22 11:58:38.000000000","message":"Patch Set 4:\n\nI tend to agree that we should turn this on by default.","accounts_in_message":[],"_revision_number":4},{"id":"984d5be2ce983ef0d992cafe5c8a914eb8965205","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-22 13:58:24.000000000","message":"Patch Set 4:\n\nI can go either way on the default settings, let\u0027s see what the wider cinder community thinks.\n\nAbout the microversioning: I\u0027m not sure how we could make that work given the current policies.\n\nSuppose the change described in this spec is mv 3.77.\nThe policy \"volume_extension:access_types_extra_specs\" is set to any authorized user (was previously admin only).\nIf a \"regular\" user requests mv 3.76, they get the full extra specs.\nIf a \"regular\" user requests mv 3.77, they get only the non-sensitive extra-specs.\nWe certainly don\u0027t want that to be the case.","accounts_in_message":[],"_revision_number":4},{"id":"a986726a62d461443218ffc5fb649d101c545d16","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-22 15:18:34.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e I can go either way on the default settings, let\u0027s see what the wider cinder community thinks.\n\u003e \n\u003e About the microversioning: I\u0027m not sure how we could make that work given the current policies.\n\u003e \n\u003e Suppose the change described in this spec is mv 3.77.\n\u003e The policy \"volume_extension:access_types_extra_specs\" is set to any authorized user (was previously admin only).\n\u003e If a \"regular\" user requests mv 3.76, they get the full extra specs.\n\u003e If a \"regular\" user requests mv 3.77, they get only the non-sensitive extra-specs.\n\u003e We certainly don\u0027t want that to be the case.\n\nRight, we wouldn\u0027t want to return any extra-specs, sensitive or not, at 3.76 or earlier.\n\nTo deal with this (and we should be more explicit in the spec) you check on the request\u0027s microversion when building the view so that at less than 3.77 you do not return *any* extra-specs to non-admin users and at 3.77 and greater you return user-visible extra-specs to non-admin users.\n\nNova and Manila already check request microversions and return different views as a function of these.","accounts_in_message":[],"_revision_number":4},{"id":"bbfe349e620425285d8171ace7401b5bf4376d0e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-22 17:48:09.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e \u003e Patch Set 4:\n\u003e \u003e \n\u003e \u003e I can go either way on the default settings, let\u0027s see what the wider cinder community thinks.\n\u003e \u003e \n\u003e \u003e About the microversioning: I\u0027m not sure how we could make that work given the current policies.\n\u003e \u003e \n\u003e \u003e Suppose the change described in this spec is mv 3.77.\n\u003e \u003e The policy \"volume_extension:access_types_extra_specs\" is set to any authorized user (was previously admin only).\n\u003e \u003e If a \"regular\" user requests mv 3.76, they get the full extra specs.\n\u003e \u003e If a \"regular\" user requests mv 3.77, they get only the non-sensitive extra-specs.\n\u003e \u003e We certainly don\u0027t want that to be the case.\n\u003e \n\u003e Right, we wouldn\u0027t want to return any extra-specs, sensitive or not, at 3.76 or earlier.\n\u003e \n\u003e To deal with this (and we should be more explicit in the spec) you check on the request\u0027s microversion when building the view so that at less than 3.77 you do not return *any* extra-specs to non-admin users and at 3.77 and greater you return user-visible extra-specs to non-admin users.\n\u003e \n\u003e Nova and Manila already check request microversions and return different views as a function of these.\n\nI should say explicitly I agree with Brian about using policy rather than context.is_admin to check whether the request is from an admin user.  Updated the prototype https://review.opendev.org/c/openstack/cinder/+/797170 accordingly.  Can put an example there of microversion check when building the API response view if that would be useful.","accounts_in_message":[],"_revision_number":4},{"id":"3955811ca461c6feff71627b76c64c1593703799","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-22 17:57:44.000000000","message":"Patch Set 4:\n\n\u003e Can put an example there of microversion check when building the API response view if that would be useful.\n\nIt\u0027s not necessary.  I understand the concept.  I just think it\u0027s pointless because it\u0027s going to give counterintuitive results for the GET /types/{type_id} and GET /types/{type_id}/{key} responses, because the new policy defaults will allow you to make them no matter what mv you\u0027re using, but if you ask for mv 3.76, you\u0027ll get an empty response.","accounts_in_message":[],"_revision_number":4},{"id":"fbf687c7503966d2da943d319a449c6c422f018b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-22 18:14:50.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e \u003e Can put an example there of microversion check when building the API response view if that would be useful.\n\u003e \n\u003e It\u0027s not necessary.  I understand the concept.  I just think it\u0027s pointless because it\u0027s going to give counterintuitive results for the GET /types/{type_id} and GET /types/{type_id}/{key} responses, because the new policy defaults will allow you to make them no matter what mv you\u0027re using, but if you ask for mv 3.76, you\u0027ll get an empty response.\n\nI realize I should have just said earlier on that if we don\u0027t need to do the microversion bump for this, then that\u0027s great :D","accounts_in_message":[],"_revision_number":4},{"id":"4e92982c48eab97fb352880c9e35a480a9770dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-22 20:00:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"353fa68d936c6c51d7aefa3b31af036b314cf3c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-22 20:05:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"622a7fa03ad9ce8c6c9d79aba15a731e0c78bf39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 20:17:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba084ed5011460d898d2379c92d2ba8 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3685bc229cd47a1a9b92649f4e57de5 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":6},{"id":"aaf555da7153da9f836bc6eee429fc6dc36d714f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-22 20:29:52.000000000","message":"Patch Set 6:\n\nThanks, Alan, I like it.  And thanks Brian for the constructive review engagement!","accounts_in_message":[],"_revision_number":6},{"id":"a721e3c9da2e4a1be26c9132c5dd60711538685a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-22 20:34:19.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6b0da2b277281319a27b3e771bd5c8930e7ded75","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-22 21:42:42.000000000","message":"Patch Set 6: Code-Review-1\n\n(10 comments)\n\nThe tables with the policies and values are very helpful in explaining how this should function.  Very nice job with those.\n\nI just have a few suggestions for clarifications noted inline; after that I\u0027m +2 on this.","accounts_in_message":[],"_revision_number":6},{"id":"2d38144a3d47810688426ad43c8e6c1653a66dc9","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-22 23:32:34.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b46c111cdfb232d4adfc44e30b6c8c61dc762d16","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-23 08:29:58.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)\n\nSorry for coming back so late. I\u0027m not exactly sure of any other extra specs that can be valid for this case however see my comment inline for encryption part.\n-1 for the wrong replication extra spec key name","accounts_in_message":[],"_revision_number":6},{"id":"3e828b97e117220cf9321365a93735a9b32e8dec","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-23 10:35:16.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f9018d281893e6f80b9c94d5bbb1bc0f9d5a0338","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 13:49:33.000000000","message":"Patch Set 6:\n\n(14 comments)","accounts_in_message":[],"_revision_number":6},{"id":"313e0499028b7cc5aae780ed9c59caecfe6f64db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 13:49:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dbfc0583982c1854c9d570f190ee12e1a6c541f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 14:01:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89a494dbda3449298a1fde59b7e988c : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c49a12ddf974cf2ba08359f644b5299 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":7},{"id":"c24457deec837cb2a73b7d71357768d6a91e3b52","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-23 14:12:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c605a6885e054e29244d6c837f3209e23d62ff13","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 14:24:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"12481c953502f45c5e1cbe6a410f0e05c54aa2af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 14:24:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bbfeb6e6d59dd60d101f13c4c15d01bea00839c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 14:33:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68445f82575a44afae80b3410edcc44f : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebdf148ddee4ca0a5b1f5229757c34f : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":8},{"id":"a2289827b0561fd1d278565db0fab0d8ff021638","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-06-23 15:20:11.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)\n\nI have always been a little surprised that users couldn\u0027t see any of the extra_specs.  I think this is an ok approach.","accounts_in_message":[],"_revision_number":8},{"id":"05c833ec3faf331f9488565b625644ee1f84b027","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 15:26:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"55b4370b8049d53d2dbe506d64e9d0090d823bf6","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 15:44:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f20cbac6c250257052a5296b7f8ce0bd4b0fe714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 15:44:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e083c823673ba8154e18b4a49494f1bf20121a93","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-23 15:47:54.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7cce0d6b01fe6f295c9b04fc1f27ba462990f3e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 16:10:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cedcf0517c544ec82bdd1e21a44d84c : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d76d510789f423bb73155eb0fc66982 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":9},{"id":"21f6c58c955a7765a8dc53d16192d1b340294b08","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-23 17:56:40.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)\n\nSorry I missed this on earlier readings, but there are 4 API calls affected (you left out GET /types).  Other than that, I\u0027m completely on board with this proposal (including changing the default values of the policies).","accounts_in_message":[],"_revision_number":9},{"id":"0874c7571c620ec74817d0bf0f244c5298862a1a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 20:56:26.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b6f0cf9a866a018094ab76b0f6f08a44d1173cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-23 20:56:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c18ec3763fc1d6a1514e0ff394fb4bcb73b07c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:09:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a6da39a6d648c9bb16e9127c9befb0 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1946553a8bd547d680d815989874cdef : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":10},{"id":"ca34d2cd3a177d27831e3bf0478646b6aa3f6dcd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-23 22:18:13.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)\n\nOne minor typo noted inline (but the intent is clear enough, so I don\u0027t think you need to revise unless you\u0027re pushing an updated patch to address something else).\n\nThe spec is clear and I think it will satisfy the use case it was introduced for without compromising the security of any backend-specific info in extra-specs.","accounts_in_message":[],"_revision_number":10},{"id":"bef4c688bdc29bf1abd05dfca8f7db305b63cd25","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-24 13:21:03.000000000","message":"Patch Set 10: Code-Review+1\n\n(5 comments)\n\nThe main thing lacking in the spec i found was the case when operators don\u0027t want to show user visible extra specs to their normal users.\nThe spec only discusses cases when we want to show all extra_specs vs user visible extra specs but not when we don\u0027t want to show any extra spec.","accounts_in_message":[],"_revision_number":10},{"id":"906e2763dca28d3bb4842f37d4b1b86f2cd324b3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-24 13:43:39.000000000","message":"Patch Set 10: Code-Review+1\n\n(3 comments)\n\nDowngrading to +1, as I agree with one of Rajat\u0027s points (though not the other; see comments inline).","accounts_in_message":[],"_revision_number":10},{"id":"304a6c50de71c3640383b5d1f816d536f2b1cc8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-24 13:50:32.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nRajat had another point about it not being obvious that operators can just turn all this off and have legacy behavior.  I re-read the spec and it\u0027s implicitly stated; suggestion inline for where it can be made more explicit.","accounts_in_message":[],"_revision_number":10},{"id":"6c2a3eb70b6a14a6456da9ae2afae04bd89e57a3","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-24 14:33:10.000000000","message":"Patch Set 10:\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d7fdaff4770ee6318b5d1d98f17e08af106b88fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-24 14:33:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"704dc8e720497a0ee43773753f2fcad8af680358","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-24 14:46:01.000000000","message":"Patch Set 11: Code-Review+2\n\nRevisions LGTM.","accounts_in_message":[],"_revision_number":11},{"id":"c417a28732f544df671ea90e4bf1baf84dce7172","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-24 14:47:23.000000000","message":"Patch Set 11: Code-Review+2\n\nLGTM, thanks Alan.","accounts_in_message":[],"_revision_number":11},{"id":"23762c7bd912ae2dff2c2f838b08c71b39630c10","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-24 15:16:43.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"12ca7f10b25291c4147c42d9386941a363d77022","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-06-24 20:01:33.000000000","message":"Patch Set 11: Code-Review+2\n\nComments have been addressed.  Looks good to me.\n\nThanks!","accounts_in_message":[],"_revision_number":11},{"id":"21f8ea93deffbcbbe540ecb011a550f08bb59124","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-25 03:30:10.000000000","message":"Patch Set 11: Workflow+1\n\nAll the points that have come up in discussion have been addressed, and the community has had sufficient time to comment on this spec.","accounts_in_message":[],"_revision_number":11},{"id":"c4357b6d6dfa9484114d09edb80624a49c52341d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-30 12:43:53.000000000","message":"Patch Set 11:\n\nLooks like this was caught up in the Zuul downtime last week.  Note that the approval date is 24 June, which was before the Cinder Xena spec freeze.","accounts_in_message":[],"_revision_number":11},{"id":"a8018de2544a0c23d9d0f2bf97ba4b117863a957","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-30 12:44:02.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0edd68db5448e9a502e35f72311b92d3fd40a1f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 12:57:03.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31456b21d2ea4d0b95db3bf3fe81f061 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf83efb7e704ee5b3f648f541c7c44d : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":11},{"id":"c063ae22b12e29ceb2737ffd06535a6332264f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 12:57:18.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1766ee9b9f4ce1b6c2c230377e3ea517c2ad5f3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:07:20.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/340505e313934af4810813ffc57f97bb : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d4c63d22f2c4616b1f53dbfe1131fa7 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":11},{"id":"91573e71d4aceae9ad5b34613bb6ebf3564a18a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:07:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"5bf5c090e608c4f08785130671e5b9c9775fcad2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:09:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/5d2ad1e705264313b7d86d6091ff95bd : SUCCESS in 59s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"7d0785fa5acc31c5310667a3af20e9cb4f1faac4","revisions":{"137eea2092e67e9d7853da8c70f75159c4d86d4a":{"kind":"REWORK","_number":1,"created":"2021-06-13 20:21:04.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/66/796166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/1"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-13 20:09:50.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-13 20:20:26.000000000","tz":-240},"subject":"Expose ``tenant visible`` extra specs","message":"Expose ``tenant visible`` extra specs\n\nto regular users so they know the abstract back end\nindependent capabilities and features of volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-tenant-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/137eea2092e67e9d7853da8c70f75159c4d86d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/137eea2092e67e9d7853da8c70f75159c4d86d4a"}]},"branch":"refs/heads/master"},"2e8ccc1a02d602bc310ed4d569ef4a2e89952fb2":{"kind":"REWORK","_number":2,"created":"2021-06-17 12:11:08.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/66/796166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/2"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-13 20:09:50.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-17 12:10:59.000000000","tz":-240},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2e8ccc1a02d602bc310ed4d569ef4a2e89952fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2e8ccc1a02d602bc310ed4d569ef4a2e89952fb2"}]},"branch":"refs/heads/master"},"51c25b607382cedb3e592f235d617f903b35e7af":{"kind":"REWORK","_number":3,"created":"2021-06-17 19:43:21.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/3"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-13 20:09:50.000000000","tz":-240},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-17 19:42:57.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/51c25b607382cedb3e592f235d617f903b35e7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/51c25b607382cedb3e592f235d617f903b35e7af"}]},"branch":"refs/heads/master"},"28ab6cf66bc7ed8b24cc894927baaf36cebb2a9f":{"kind":"REWORK","_number":4,"created":"2021-06-21 18:21:05.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/4"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-21 18:20:57.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-21 18:20:57.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/28ab6cf66bc7ed8b24cc894927baaf36cebb2a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/28ab6cf66bc7ed8b24cc894927baaf36cebb2a9f"}]},"branch":"refs/heads/master"},"9e4c28ef37b435d663c83bab39e86fe2921c1aed":{"kind":"REWORK","_number":5,"created":"2021-06-22 20:00:24.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/5"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-21 18:20:57.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-22 20:00:16.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9e4c28ef37b435d663c83bab39e86fe2921c1aed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9e4c28ef37b435d663c83bab39e86fe2921c1aed"}]},"branch":"refs/heads/master"},"e6fe4c5c6204f88cac34e6b9be5d6835266e5161":{"kind":"REWORK","_number":6,"created":"2021-06-22 20:05:38.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/6"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-21 18:20:57.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-22 20:05:29.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e6fe4c5c6204f88cac34e6b9be5d6835266e5161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e6fe4c5c6204f88cac34e6b9be5d6835266e5161"}]},"branch":"refs/heads/master"},"f07858538e32b0f830c01e7386ad2e4141156c98":{"kind":"REWORK","_number":7,"created":"2021-06-23 13:49:39.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/7"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-23 13:45:04.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-23 13:49:15.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f07858538e32b0f830c01e7386ad2e4141156c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f07858538e32b0f830c01e7386ad2e4141156c98"}]},"branch":"refs/heads/master"},"0872da1c80470aa526a40b4df0f555ff31e20aa4":{"kind":"REWORK","_number":8,"created":"2021-06-23 14:24:46.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/8"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-23 13:45:04.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-23 14:23:50.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0872da1c80470aa526a40b4df0f555ff31e20aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0872da1c80470aa526a40b4df0f555ff31e20aa4"}]},"branch":"refs/heads/master"},"afb6614b003b99d21913f22a1bbd1b83b3c8fcc7":{"kind":"REWORK","_number":9,"created":"2021-06-23 15:44:53.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/9"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-23 13:45:04.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-23 15:44:08.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/afb6614b003b99d21913f22a1bbd1b83b3c8fcc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/afb6614b003b99d21913f22a1bbd1b83b3c8fcc7"}]},"branch":"refs/heads/master"},"b5defe68465847085617f23d67c8938d58e23603":{"kind":"REWORK","_number":10,"created":"2021-06-23 20:56:33.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/10"}}},"commit":{"parents":[{"commit":"eff52b7353fe8fbe14ebd70ab455c76ceb71c702","subject":"Merge \"Snapshot attached volumes w/o force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/eff52b7353fe8fbe14ebd70ab455c76ceb71c702"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-23 20:52:41.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-23 20:55:37.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b5defe68465847085617f23d67c8938d58e23603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b5defe68465847085617f23d67c8938d58e23603"}]},"branch":"refs/heads/master"},"7d0785fa5acc31c5310667a3af20e9cb4f1faac4":{"kind":"REWORK","_number":11,"created":"2021-06-24 14:33:22.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/66/796166/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/796166/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/796166/11"}}},"commit":{"parents":[{"commit":"eff52b7353fe8fbe14ebd70ab455c76ceb71c702","subject":"Merge \"Snapshot attached volumes w/o force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/eff52b7353fe8fbe14ebd70ab455c76ceb71c702"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-06-24 14:32:49.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2021-06-24 14:32:49.000000000","tz":-420},"subject":"Expose ``user visible`` extra specs","message":"Expose ``user visible`` extra specs\n\nso that regular users can know the abstract back end\nindependent capabilities and features in volume types.\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/expose-user-visible-extra-specs\n\nAPIImpact\nDocImpact\n\nCo-Authored-By: Tom Barron \u003ctpb@dyncloud.net\u003e\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\n\nChange-Id: Id236326dca9ee08aad8427ea6cb400e25e7ac09e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7d0785fa5acc31c5310667a3af20e9cb4f1faac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7d0785fa5acc31c5310667a3af20e9cb4f1faac4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
