)]}'
{"id":"openstack%2Fcinder-specs~796365","triplet_id":"openstack%2Fcinder-specs~master~I94c1e8df684a25b0630af6490966b376dad6c047","project":"openstack/cinder-specs","branch":"master","topic":"bp/nvmeof-connection-agent","hashtags":[],"change_id":"I94c1e8df684a25b0630af6490966b376dad6c047","subject":"NVMe-oF connection agent","status":"MERGED","created":"2021-06-15 05:04:21.000000000","updated":"2021-06-26 13:55:50.000000000","submitted":"2021-06-25 20:58:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"796365-bp/nvmeof-connection-agent","meta_rev_id":"afc5f440b7c474999bda319fdc91cfc74e7328de","_number":796365,"virtual_id_number":796365,"owner":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-25 20:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-06-25 03:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-24 21:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-06-25 09:26:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-06-25 20:50:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 05:13:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-23 02:47:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-06-23 14:11:12.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-06-23 14:50:55.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-06-24 21:40:56.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"0268e63c7913746597b33330faa47f244d21185b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-15 05:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d7c10a950ecc8555263bd0852fc9ae4b81c110c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 05:13:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7d725f44c04ef1be39d1a860884912 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c9aa63992d4b8fbc2658a398e47e1f : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"34b8cd461f4a760867678b0e5db12c85f7d04518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-15 09:28:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"955837fef549d61aaf82513cf708ddcd22ba2602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 09:38:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/637330a0693544dea4da2483602bb625 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6fc393b2c44f50bde7d0ba3df1a304 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"88c3fe7a115e213524bd904e7c3fec56f3a1401a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-21 07:31:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"86196e36c47c200b8c10fe836cc5bd67f92d3ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 07:39:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9562b3e1adc2453287ab980d735ae63a : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c3f6241e974b86b471ee4faeea0a8b : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"cb5cbb3abeb7917f318afb1e2f43263f4219f4dd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-23 02:47:19.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nA point you could clarify noted inline.  Otherwise, this seems to capture what was discussed at the  PTG.  (The -1 is just to get your attention; I agree with the overall proposal.)","accounts_in_message":[],"_revision_number":3},{"id":"245bb60fc8d9fd48c64fc8b97f849d3e312e433f","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-23 08:48:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0476a45c43184f9f6d242d131ea31232ace72856","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-23 08:49:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"828988edf643e930643800bfd8193df38cac4a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 08:55:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d585b07d58864d12abaac6ab918c4c98 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db8799e3e5c43b78f6b00bac58753c6 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"ec22aa7e00eca2f0dea2c355f4a8a9a4d12c0433","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-23 12:46:47.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)\n\nZohar: the revision looks good, two nits noted inline that you can correct if you have to make other revisions.  (Only a +1 because I want to read through the entire spec one more time.)","accounts_in_message":[],"_revision_number":4},{"id":"d46c723a9ddaabb32cc93425f3f8c84bf30a9bcf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-23 14:11:12.000000000","message":"Patch Set 4: Code-Review+1\n\n(7 comments)\n\nFew comments inline (mostly queries) and suggestions which are based on my perspective of reading the spec.","accounts_in_message":[],"_revision_number":4},{"id":"b35209d3d81e9c0df6c3df38cf5407875e89938a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-06-23 14:50:55.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nI am ok with this.  I would like to see a +2 from people more intimately involved with os-brick.","accounts_in_message":[],"_revision_number":4},{"id":"d983a6ff4290a23c1e221b9715ef8c06acde5163","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-24 04:53:24.000000000","message":"Patch Set 4:\n\n(10 comments)\n\nThank you all for the great comments!\n\nThere are some of them I left \"unresolved\" - but they are mostly about readability and explicitness (lines 34, 49 and 130), open discussions (lines 59 and 99), or minor techincal detail that is open for more input (line 77)","accounts_in_message":[],"_revision_number":4},{"id":"673339d28f64945f8615506ed5a941e05de8f662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-24 04:53:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"371b2bcc427e41e42ff326b6562042fc7e284a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 05:03:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1c54813a984d9f843ed2332afb4adc : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac35c6414c1451a910705ae18df5375 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":5},{"id":"86c849145752f158be3d8996bbf15f6a5a06463c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-24 12:23:37.000000000","message":"Patch Set 5: Code-Review+2\n\n(4 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"082317cd7c9cacfa8f189fa1f0c089ce1d676d66","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-24 15:06:06.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nThank you Rajat, after deeper understanding the glance usecase with your help, i reworded it a bit to fit overall understanding.","accounts_in_message":[],"_revision_number":5},{"id":"d05fdaaa5c61c7acad3e824f334c3b9dcd726a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-24 15:06:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0d5ffcc8ae8163505abf7e39aebfa0bea7bebafb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 15:18:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e6b31cd81740b4ba0a0226d05c53b5 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270d7c8635c14f898c804dc61eceea85 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":6},{"id":"b0f8f3abfc37c29eef9de148093cce2471a06363","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-24 21:40:56.000000000","message":"Patch Set 6: Code-Review+1\n\nI think It would have been nice to explicitly mention the Glance case that Rajat pointed out in but I\u0027m not so familiar with this +1","accounts_in_message":[],"_revision_number":6},{"id":"bddc87c27d1f392ceb063be35ec168aad13770b5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-25 03:28:13.000000000","message":"Patch Set 6: Code-Review+2\n\nI think it\u0027s pretty clear what you\u0027re proposing to implement, and I think you\u0027ve addressed all the major points that came up in discussion at the PTG and on this review.","accounts_in_message":[],"_revision_number":6},{"id":"283a3c553e713566d361f6be180583a7f6181ca6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-25 09:26:30.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0935b354014d6ecbce4935802a9b310056f64536","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-25 20:50:55.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)\n\nI think everyone\u0027s had sufficient time to comment on the spec.  An issue occurred to me that isn\u0027t mentioned in the spec, but I\u0027m not going to hold it up.  Please make sure you read the comment inline, though.","accounts_in_message":[],"_revision_number":6},{"id":"35cb28f0fceed6f8e39766c0494a678b9deacdd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 20:51:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"acf210223ce73e26b18f121cc2f08d97f0e75497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 20:58:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f896bb54134ce59dd899fa48afde3d : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2797e5ef63384c48a16ff07e269639b1 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":6},{"id":"1aa432c9b099f0b06ee5f1d5360d3ffb2e15899b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 20:58:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"1cbd25ac3bdb9369122c6a8fef486bf05582ba7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 20:59:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/fdb0fa21d5ae49ffa77be909be491f61 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":6},{"id":"afc5f440b7c474999bda319fdc91cfc74e7328de","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-06-26 13:55:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"46bfe10ef67279a6e265d5b7a4d25dfd6c701154","revisions":{"b1530705e2e704cc66af09a8f0f564e026b55f38":{"kind":"REWORK","_number":1,"created":"2021-06-15 05:04:21.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/65/796365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/65/796365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/1"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 05:03:37.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 05:03:37.000000000","tz":0},"subject":"Implements: blueprint nvmeof-client-raid-healing-agent","message":"Implements: blueprint nvmeof-client-raid-healing-agent\n\nChange-Id: I94c1e8df684a25b0630af6490966b376dad6c047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b1530705e2e704cc66af09a8f0f564e026b55f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b1530705e2e704cc66af09a8f0f564e026b55f38"}]},"branch":"refs/heads/master"},"192d6a1c35b8eb8d22b93bdac4a500006ac6c89f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-15 09:28:17.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/65/796365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/65/796365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/2"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 05:03:37.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 09:26:40.000000000","tz":0},"subject":"NVMe-oF connection agent","message":"NVMe-oF connection agent\n\nImplements: blueprint nvmeof-connection-agent\n\nChange-Id: I94c1e8df684a25b0630af6490966b376dad6c047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/192d6a1c35b8eb8d22b93bdac4a500006ac6c89f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/192d6a1c35b8eb8d22b93bdac4a500006ac6c89f"}]},"branch":"refs/heads/master"},"b50276653f34938d615c084a9dc57c8615d0e51f":{"kind":"REWORK","_number":3,"created":"2021-06-21 07:31:59.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/65/796365/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/65/796365/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/3"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 05:03:37.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-21 07:31:24.000000000","tz":0},"subject":"NVMe-oF connection agent","message":"NVMe-oF connection agent\n\nImplements: blueprint nvmeof-connection-agent\n\nChange-Id: I94c1e8df684a25b0630af6490966b376dad6c047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b50276653f34938d615c084a9dc57c8615d0e51f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b50276653f34938d615c084a9dc57c8615d0e51f"}]},"branch":"refs/heads/master"},"1dae425b7ab769464082b11d507d2a761df455cd":{"kind":"REWORK","_number":4,"created":"2021-06-23 08:49:08.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/65/796365/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/65/796365/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/4"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 05:03:37.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-23 08:48:20.000000000","tz":0},"subject":"NVMe-oF connection agent","message":"NVMe-oF connection agent\n\nImplements: blueprint nvmeof-connection-agent\n\nChange-Id: I94c1e8df684a25b0630af6490966b376dad6c047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1dae425b7ab769464082b11d507d2a761df455cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1dae425b7ab769464082b11d507d2a761df455cd"}]},"branch":"refs/heads/master"},"0dd6a51ec04097711c3091e5026add9ab3981fb2":{"kind":"REWORK","_number":5,"created":"2021-06-24 04:53:45.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/65/796365/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/65/796365/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/5"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 05:03:37.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-24 04:53:15.000000000","tz":0},"subject":"NVMe-oF connection agent","message":"NVMe-oF connection agent\n\nImplements: blueprint nvmeof-connection-agent\n\nChange-Id: I94c1e8df684a25b0630af6490966b376dad6c047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0dd6a51ec04097711c3091e5026add9ab3981fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0dd6a51ec04097711c3091e5026add9ab3981fb2"}]},"branch":"refs/heads/master"},"46bfe10ef67279a6e265d5b7a4d25dfd6c701154":{"kind":"REWORK","_number":6,"created":"2021-06-24 15:06:34.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/65/796365/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/65/796365/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/65/796365/6"}}},"commit":{"parents":[{"commit":"e2abf973c7e4e80293aad403a10b2455ca34b573","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/e2abf973c7e4e80293aad403a10b2455ca34b573"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-15 05:03:37.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-06-24 15:05:58.000000000","tz":0},"subject":"NVMe-oF connection agent","message":"NVMe-oF connection agent\n\nImplements: blueprint nvmeof-connection-agent\n\nChange-Id: I94c1e8df684a25b0630af6490966b376dad6c047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/46bfe10ef67279a6e265d5b7a4d25dfd6c701154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/46bfe10ef67279a6e265d5b7a4d25dfd6c701154"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
