)]}'
{"id":"openstack%2Fcinder-specs~809622","triplet_id":"openstack%2Fcinder-specs~master~I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509","project":"openstack/cinder-specs","branch":"master","topic":"bp/add-volume-re-image-api","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2021-12-19 15:54:19.000000000","reason":"Brian Rosmaita replied on the change"},"10135":{"account":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"last_update":"2021-12-21 20:47:11.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2021-12-21 20:47:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509","subject":"Add volume reimage spec","status":"MERGED","created":"2021-09-17 09:51:06.000000000","updated":"2021-12-21 20:48:52.000000000","submitted":"2021-12-21 20:47:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"809622-bp/add-volume-re-image-api","meta_rev_id":"d20d76c2997d60f324205a98a68733c0967962ea","_number":809622,"virtual_id_number":809622,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 20:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-12-19 15:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-21 20:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-21 20:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 10:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-10 13:57:43.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-12-07 17:41:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-12-07 18:59:51.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-12-21 20:39:34.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"960169ec232a0d779d40530218d6065b25dfc3e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-17 09:51:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e90dfe414ec55dc0313e5c70421f556203f74b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-17 09:57:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12780c0a0aa6870956b9344877f3350ee8a56594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:13:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0b80a45186480391cbc9c2bdc00d87 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361257a700a3437883adb06bfce74969 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"430e88f32ad36558ac40d18add39cae20b9c58be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-10-22 08:52:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b129865f528c26198fb547435191770068efd7e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 09:14:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bb5c2c930f460d8a7bbfb390fd9fbf : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b176bb252bc14ef7b7ef922548f87bc0 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"0dec2f5bb58e4f1c2ac524abee3ae3de75804924","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-10 13:57:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d280fe8340bb15fbf00414c98476e3e814b3f060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-11 08:25:30.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6df9064f1fc48024ac282fe7a3f8c6a3a924a651","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-11 08:30:19.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"df7926c59b85953d1637ff3aec8b5f1247981bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 08:35:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b329f253c289428ca819f1c83f7fd30c : FAILURE in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac57d6e77dec48e888445cc2012aa47a : SUCCESS in 4m 15s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/add-volume-re-image-api.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/add-volume-re-image-api.rst","accounts_in_message":[],"_revision_number":4},{"id":"bbc2ff7018f4ee2efdfa72bdc1a46e20f035b65a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-11 09:43:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c9bd00ec26d83c17565b77efd54b48a9e3fd113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 09:54:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875c6d4d9a1f4f80ac26487b37597c2d : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d1779d92384ac88b940bf3f81e7f4c : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":5},{"id":"af73474aee398bd75d52bb89fc8a2f10f0bd9d8a","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-17 11:24:10.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ea1cfb99b366b697228043ad7b5464dd1f813006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-19 08:59:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f39908b76b7da0a1ac71d2dc226a8faf92472ae8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-19 09:05:08.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1452c9a89e80696cf453e8f87ad9d6d544cc0248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 09:10:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcc01924b1c42a99a9797d78dc62184 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03369fa9c15e49fe87c7b36e6bcddd7c : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":6},{"id":"9b1be815fe5906cd4c0a8e0661325f10418e244c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-07 17:41:55.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5a6714801967c1d2358604bbdbe508aefa8c5224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-12-10 10:49:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"01893285fb42b7701796202b9c0d0608a0cb3efe","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-12-10 10:53:35.000000000","message":"Patch Set 7:\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"350f33e1086f17473cd316a32230568e8297f3df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 10:58:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4040777557445988814c5ad9156ab57 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da39905d9ff34b52ae469b484b7a2b4d : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":7},{"id":"95d07d57cc300b508f210317c20b3a69cb5fbb35","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-19 15:54:19.000000000","message":"Patch Set 7: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3204f15c8350c39bd67df521b4a8434e03d2b743","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-12-21 20:39:34.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4babef36f4b63e057fbc210553cbf388c5204f17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 20:40:04.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"79d18bd58c3ca7c9f094a6ab5806b1f7ccc2eca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 20:47:11.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5a375d831f46309c656c478277d9e0 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3161788854a248c4a6f3ae583c0f8cce : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":7},{"id":"860bc1e63fa723664f202a57b6289e7c0829f0ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 20:47:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"d20d76c2997d60f324205a98a68733c0967962ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 20:48:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/ce99898954144618a64e733b3fbcf4b6 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a5e710f8ba6cddf21dbc535d84e2b5792ade236c","revisions":{"3b62336f075d4c129962a7610dd15cd4bba3d03a":{"kind":"REWORK","_number":1,"created":"2021-09-17 09:51:06.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/809622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/809622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/1"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"subject":"Add volume reimage spec","message":"Add volume reimage spec\n\nThis patch adds the spec for reimage feature which enables the\nfunctionality to rebuild a volume backed instance.\n\nImplements: blueprint add-volume-re-image-api\nChange-Id: I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b62336f075d4c129962a7610dd15cd4bba3d03a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3b62336f075d4c129962a7610dd15cd4bba3d03a"}]},"branch":"refs/heads/master"},"df73490c2683682f1f9f11ce9407262079feb0aa":{"kind":"REWORK","_number":2,"created":"2021-09-17 09:57:33.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/809622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/809622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/2"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:55:41.000000000","tz":-240},"subject":"Add volume reimage spec","message":"Add volume reimage spec\n\nThis patch adds the spec for reimage feature which enables the\nfunctionality to rebuild a volume backed instance.\n\nImplements: blueprint add-volume-re-image-api\n\nChange-Id: I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/df73490c2683682f1f9f11ce9407262079feb0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/df73490c2683682f1f9f11ce9407262079feb0aa"}]},"branch":"refs/heads/master"},"0d0321da0f6987b63941fb0d7e421f63dc4ec95e":{"kind":"REWORK","_number":3,"created":"2021-10-22 08:52:04.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/809622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/809622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/3"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-10-22 08:47:29.000000000","tz":-240},"subject":"Add volume reimage spec","message":"Add volume reimage spec\n\nThis patch adds the spec for reimage feature which enables the\nfunctionality to rebuild a volume backed instance.\n\nImplements: blueprint add-volume-re-image-api\n\nChange-Id: I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0d0321da0f6987b63941fb0d7e421f63dc4ec95e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0d0321da0f6987b63941fb0d7e421f63dc4ec95e"}]},"branch":"refs/heads/master"},"d419253217186676013b0efc39a1c321d2276f34":{"kind":"REWORK","_number":4,"created":"2021-11-11 08:25:30.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/809622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/809622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/4"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-11-11 08:25:30.000000000","tz":0},"subject":"Add volume reimage spec","message":"Add volume reimage spec\n\nThis patch adds the spec for reimage feature which enables the\nfunctionality to rebuild a volume backed instance.\n\nImplements: blueprint add-volume-re-image-api\n\nChange-Id: I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d419253217186676013b0efc39a1c321d2276f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d419253217186676013b0efc39a1c321d2276f34"}]},"branch":"refs/heads/master"},"7294eb863858d4ec8b1c88ae7c2a0aa0755f3212":{"kind":"REWORK","_number":5,"created":"2021-11-11 09:43:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/809622/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/809622/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/5"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-11 09:43:18.000000000","tz":-300},"subject":"Add volume reimage spec","message":"Add volume reimage spec\n\nThis patch adds the spec for reimage feature which enables the\nfunctionality to rebuild a volume backed instance.\n\nImplements: blueprint add-volume-re-image-api\n\nChange-Id: I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7294eb863858d4ec8b1c88ae7c2a0aa0755f3212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7294eb863858d4ec8b1c88ae7c2a0aa0755f3212"}]},"branch":"refs/heads/master"},"159f6effd43a2307b185bee97e0423cf6822de00":{"kind":"REWORK","_number":6,"created":"2021-11-19 08:59:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/809622/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/809622/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/6"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-19 08:58:55.000000000","tz":-300},"subject":"Add volume reimage spec","message":"Add volume reimage spec\n\nThis patch adds the spec for reimage feature which enables the\nfunctionality to rebuild a volume backed instance.\n\nImplements: blueprint add-volume-re-image-api\n\nChange-Id: I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/159f6effd43a2307b185bee97e0423cf6822de00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/159f6effd43a2307b185bee97e0423cf6822de00"}]},"branch":"refs/heads/master"},"a5e710f8ba6cddf21dbc535d84e2b5792ade236c":{"kind":"REWORK","_number":7,"created":"2021-12-10 10:49:20.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/22/809622/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/809622/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/809622/7"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-09-17 09:48:07.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-12-10 10:49:05.000000000","tz":-300},"subject":"Add volume reimage spec","message":"Add volume reimage spec\n\nThis patch adds the spec for reimage feature which enables the\nfunctionality to rebuild a volume backed instance.\n\nImplements: blueprint add-volume-re-image-api\n\nChange-Id: I067d113cacaf9a4d9e6a7f4d7ed1e0dfcc3ae509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a5e710f8ba6cddf21dbc535d84e2b5792ade236c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a5e710f8ba6cddf21dbc535d84e2b5792ade236c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
