)]}'
{"id":"openstack%2Fcinder-specs~809741","triplet_id":"openstack%2Fcinder-specs~master~If6c289d6957bd739ed893f01519441e6039a7d00","project":"openstack/cinder-specs","branch":"master","topic":"update-s-rbac-specs","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2021-12-17 19:45:00.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2021-12-17 19:36:24.000000000","reason":"Jay Bryant replied on the change"}},"hashtags":[],"change_id":"If6c289d6957bd739ed893f01519441e6039a7d00","subject":"Update \"Secure RBAC Ready\" spec","status":"MERGED","created":"2021-09-17 14:23:15.000000000","updated":"2021-12-17 19:46:41.000000000","submitted":"2021-12-17 19:45:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"809741-update-s-rbac-specs","meta_rev_id":"44407562d51da1d424733bcc99c892666e1a57a0","_number":809741,"virtual_id_number":809741,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-17 19:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-17 19:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-17 19:36:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 14:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 14:56:34.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"0c23c3e9b837145508d1b3ca452228e6c94f4779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-17 14:23:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34f91d9c351e63b7b147010d5da8b713fcfb635f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:45:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffbe912d3f564b0bb0596e0d0f3b35ae : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/794e6de2618147a4bb6d62278ccf798e : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9a615e81d7a2f96eb59ecb38b89eac8559d5adff","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-09-17 14:56:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nOne thing that needs to be addressed.  Not sure if your sentence was cut off or if you just missed a capital.","accounts_in_message":[],"_revision_number":1},{"id":"f16ef2cb8883a7eeda1a7e5a13a9c9a00a638a7b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-17 23:07:06.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for the review, Jay.  New PS up in a minute.","accounts_in_message":[],"_revision_number":1},{"id":"4a1315b1320e01a2c6d55770fe40247688f6d7da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-17 23:08:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88ffe2b8bb2f94176c0fe21aba74ee4416b14315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 23:18:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927ab2b53afb49658a5f09dd638c0f26 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca4118cef56b4925b318a0c3ea90097e : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1de5b052a3851657ba38c1f51403a4cd2e464f53","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-10-13 13:40:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"31ed4465157b86211a373e5f83b85105229739c9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-04 13:34:24.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94effa31703129df28ebbdfb1039540fbb0a43bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-17 02:54:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89d7a46ba94e21e04bf95996e5d731c15eea15cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 03:08:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a3bfb39c5847a4a241500b0134be20 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c994acfe714ee8a6d1fb16ba371e05 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"c4a0c46b7ed27b3e5bfa12ea1a7685040b10511c","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-12-17 19:36:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c7d579557c214b6e58e4c66e79e2194bfd5b0d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 19:36:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c55ceb94e107edb44a5280582cd8bbdf47902da4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 19:44:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5746043d50b4437a652c591e2e19fee : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd628398ef7f4251bb110e68c55d9437 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f19142d1c69f2ccf1337ca44f73b23e9c9d1720f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 19:45:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"44407562d51da1d424733bcc99c892666e1a57a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 19:46:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e76ea83f780543df96b185c0330c707a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"de4308f4a756da3c189ea77663715ae16a5ea215","revisions":{"f091003966e4d9da047803821a6732c2c8370b04":{"kind":"REWORK","_number":1,"created":"2021-09-17 14:23:15.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/809741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/41/809741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/1"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-09-17 14:07:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-09-17 14:21:53.000000000","tz":-240},"subject":"Update \"Secure RBAC Ready\" spec","message":"Update \"Secure RBAC Ready\" spec\n\nThe initial spec was a bit optimistic about how much would be\naccomplished in Xena and this has become a two-cycle effort.\nUpdate the Xena spec to reflect this, and add a new spec for the\ncontinuation of this work in Yoga.\n\nChange-Id: If6c289d6957bd739ed893f01519441e6039a7d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f091003966e4d9da047803821a6732c2c8370b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f091003966e4d9da047803821a6732c2c8370b04"}]},"branch":"refs/heads/master"},"5f08fc21dde26719b37e078004d3d4371496c783":{"kind":"REWORK","_number":2,"created":"2021-09-17 23:08:29.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/809741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/41/809741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/2"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-09-17 14:07:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-09-17 23:07:36.000000000","tz":-240},"subject":"Update \"Secure RBAC Ready\" spec","message":"Update \"Secure RBAC Ready\" spec\n\nThe initial spec was a bit optimistic about how much would be\naccomplished in Xena and this has become a two-cycle effort.\nUpdate the Xena spec to reflect this, and add a new spec for the\ncontinuation of this work in Yoga.\n\nChange-Id: If6c289d6957bd739ed893f01519441e6039a7d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5f08fc21dde26719b37e078004d3d4371496c783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5f08fc21dde26719b37e078004d3d4371496c783"}]},"branch":"refs/heads/master"},"de4308f4a756da3c189ea77663715ae16a5ea215":{"kind":"REWORK","_number":3,"created":"2021-12-17 02:54:38.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/809741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/41/809741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/41/809741/3"}}},"commit":{"parents":[{"commit":"06c53c4935aab4ecc27bacfb4b13757d5488a248","subject":"Merge \"Propose image-encryption spec for Yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/06c53c4935aab4ecc27bacfb4b13757d5488a248"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-09-17 14:07:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-12-17 02:53:37.000000000","tz":-300},"subject":"Update \"Secure RBAC Ready\" spec","message":"Update \"Secure RBAC Ready\" spec\n\nThe initial spec was a bit optimistic about how much would be\naccomplished in Xena.  Additionally, this initiative has been\nformalized as a multi-cycle community goal and the direction of\nthe effort has been revised and expanded.  So update the Xena\nspec to reflect what was actually done in Xena, and add a new\nspec for the continuation of this work in Yoga.\n\nChange-Id: If6c289d6957bd739ed893f01519441e6039a7d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/de4308f4a756da3c189ea77663715ae16a5ea215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/de4308f4a756da3c189ea77663715ae16a5ea215"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
