)]}'
{"id":"openstack%2Fcinder-specs~821372","triplet_id":"openstack%2Fcinder-specs~master~I9fd0700123813b1f7afc4e86177c4c67f43caaae","project":"openstack/cinder-specs","branch":"master","topic":"change-821372","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2022-03-03 17:37:52.000000000","reason":"Simon Dodsley replied on the change"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-20 19:09:01.000000000","reason":"Change was abandoned"},"24005":{"account":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"last_update":"2022-06-20 19:09:01.000000000","reason":"Change was abandoned"},"14806":{"account":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"last_update":"2022-01-25 04:32:58.000000000","reason":"Chhavi Agarwal replied on the change"},"32345":{"account":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"last_update":"2022-06-20 19:09:01.000000000","reason":"Change was abandoned"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-06-20 19:09:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9fd0700123813b1f7afc4e86177c4c67f43caaae","subject":"Add multiple volume groups - Yoga","status":"ABANDONED","created":"2021-12-10 08:25:04.000000000","updated":"2022-06-20 19:09:01.000000000","total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b23fc44883ea25db0decfddc4cf6c837188fcba","_number":821372,"virtual_id_number":821372,"owner":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-25 17:02:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":-1,"date":"2022-03-03 17:37:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},{"_account_id":26402,"name":"Kumar Prashant","email":"kumar.prashant@dell.com","username":"prashk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 08:25:35.000000000","updated_by":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"reviewer":{"_account_id":26402,"name":"Kumar Prashant","email":"kumar.prashant@dell.com","username":"prashk"},"state":"CC"},{"updated":"2021-12-10 08:25:35.000000000","updated_by":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"},{"updated":"2021-12-10 08:32:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-14 04:23:13.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-01-18 17:43:29.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2022-01-24 19:14:16.000000000","updated_by":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"reviewer":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"state":"CC"},{"updated":"2022-01-25 15:05:18.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-02-02 14:31:53.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"b7151df67e6c3422c470f2db92f03a1e922bf772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-10 08:25:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca8fd595900b9651e231490f3ff71e43a2376a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 08:32:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256cfa8dc2b147228960aa8ceda7f333 : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5dd3a1fd3a4293ba7d20ac87a2521d : FAILURE in 4m 16s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/multiple-volume-groups.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/multiple-volume-groups.rst","accounts_in_message":[],"_revision_number":1},{"id":"00a840068300f34c4c346c431f05c3edfb694970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-10 10:21:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"65eb496b20bb033793447aba8386e13ba1e8af9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 10:36:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb57855af98f4eed90c4b39c65ad7be2 : FAILURE in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc28cb102fd4b9f865b9c0a10198868 : FAILURE in 3m 30s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/multiple-volume-groups.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/multiple-volume-groups.rst","accounts_in_message":[],"_revision_number":2},{"id":"91c2f208623ccfb7f46061c92ed75b2173a9fd76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-10 11:10:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"abf1ff6fa331cb2f1d09674c0ec960170de3aa15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 11:16:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce6133b23964d0699ed33eec249e9a9 : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bfff785fb34c98bb9490e343414eda : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"657a6283d0615341a584291f3af48ef81cfb6b0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-10 11:25:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"96f323cd0f61f23b1bcb7c498d84e1135509f8b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 11:35:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f342944f0e417e87fb798ec45ecf43 : FAILURE in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59736f1b7e3440079b932b522271990c : SUCCESS in 4m 32s\n\nWarning:\n  Comments left for invalid file doc/source/index.rst","accounts_in_message":[],"_revision_number":4},{"id":"c6876d787db0d672f529e00ec62fa3f78b710934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-10 11:49:02.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a67471bf78e6ea9b6c8deee644e3a702e496247d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 11:55:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eea9aae66114232a5f7b1b86969d0bf : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223a128d981b42b9ae9f2d0f6de36e6c : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":5},{"id":"88d13d037b8c2912f6ad1baf6262184208b1bc8c","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-13 11:34:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ccadb846c0246bc87507665c51791d9f05109d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-13 11:38:03.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"9f51f7d09ab4ae0171e102ecbdfe3046beb7c929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-13 11:39:25.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"201ebe1e12e13713f8e68ad1e025d689873b9f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 11:47:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc38e879e8684953ab4932258cf93970 : FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc3ede1f20cd4cc2b65389bb04d26c6a : SUCCESS in 5m 22s\n\nWarning:\n  Comments left for invalid file doc/source/index.rst","accounts_in_message":[],"_revision_number":7},{"id":"5fadad1ca0d264d8668d4501654eef71e807eba7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-14 04:23:13.000000000","message":"Patch Set 7: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"715733444b3e3e1bd175ace9fe285acda781cce8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"date":"2021-12-16 05:45:40.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"f8ca9e77f062a92001884daa90229d6dfd6ff7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 05:55:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8ff434cbb04666b242ef7569232646 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d16df3a8d14c7d9f9ac222851a31ee : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":8},{"id":"8b859a1fef9c15cefc7e6be08b46e8a4202ddcfe","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-19 22:20:23.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bb226dc5f741cc60ab7d16cced4186c334693632","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2022-01-16 13:10:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4471d42c2e17c1bca6e30c1b1477677e9af0c686","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2022-01-16 13:55:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b0ee03b8374fb2c4c242075a908b019bf9a08879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-16 14:01:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51c0f1de587492fb29d9d63461595b9 : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7075462264a4ab48ad0091d49ac2a79 : FAILURE in 3m 06s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/multiple-volume-groups.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/multiple-volume-groups.rst","accounts_in_message":[],"_revision_number":9},{"id":"bd055c33ad3b388fcda0a9bfa7d1312423274c56","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-01-18 17:43:29.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9f136632e88b134a7acd96310268edcbba560109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2022-01-23 17:54:55.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"50cb53c8f8d5c747fae546e87c78acc42eaba0f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-23 18:03:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/599787212a544356bbfeb0378e59ff59 : FAILURE in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3fc3e3563c492e8a3a15d6e78e50e9 : FAILURE in 3m 35s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":10},{"id":"726bbf307546436ab978b4a781b83e2dad52ab86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 08:47:02.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"f04cb00902be69bcb60823e49bf9b07c50fcb32d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 08:53:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806b6b53b2024ce9acb710932d06691b : FAILURE in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3c20211ff84324a3d5ac77fd0213f1 : FAILURE in 4m 05s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":11},{"id":"ef34a3602bbc9ac2571e5d45e5294b1d107d1d9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 11:03:46.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"def7c522524a219483a334e1c328d73215d01b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:12:10.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07d7da3c83d487c9a168bfe5d4102a2 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df9b0aebf4b4efcac4df9e9ea42c672 : FAILURE in 4m 25s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":12},{"id":"8af1624d0eb0c645bd9389c9948096a36f8a4ce2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 11:23:52.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"3891e47534ead3b2a1f2e2236a003e71224accc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:29:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6fd077277c424da295e6600e28f540 : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4984b38f87b34e8f9e977785ea5f035c : FAILURE in 3m 37s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":13},{"id":"7e2845d9067b5c6ade2cd7348773e9bda4d08b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 11:39:44.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"db8a96f15276612cc35c9d85fd924ec743ccdc92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:50:19.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b528dcaecb4d0a99be6a3af42875ff : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2b6ba953b7407f8aef828a29bd6ca7 : FAILURE in 3m 47s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":14},{"id":"ce34cee65e2e26b1559f54e53cfe8bd1e92c118e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 11:57:17.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"6a0d996690b88a8e44f0cc295cc6d80b2defb364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 12:08:27.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcc7ede3b8e4bf9bbc3233e22102210 : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92230aae82c34741b0d293548a14d976 : FAILURE in 3m 35s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":15},{"id":"70fdefc0d5f8dd06f0c96eb4172f56b9c9302d2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 14:09:25.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"351fc0d7d02fbd676ffcd1e0f2d57529acd9b50a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:21:18.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec151330f65f4be1a3cbf17e5f318604 : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9960ec91bc46218805b1ac1b57cf27 : FAILURE in 4m 26s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":16},{"id":"938ec6f8060a126f618d0993c8f79e1f5303f433","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 14:34:44.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"e54cb9911622e8326c4ef30f19395bf878d453d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:40:32.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5581adf76c44a0fa303ad7f5266a7be : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd09309e03c2470a88f7b806a1a642cb : SUCCESS in 3m 56s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":17},{"id":"ed123f00ed06d64040b09b823f1e2b0c1f93136c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 15:33:19.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"9f7836e47f722bc214dea27fa55f0237d8b06dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 15:37:38.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"a8c319ae334049e3058c105146f5b54866c5ddda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 15:49:52.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/235b3bbc3a674ae4852b20333fb0a3f5 : FAILURE in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb964896faf4db38c110e3dc238e15e : SUCCESS in 3m 50s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":19},{"id":"8c8330b0bfb91f9c8055cd9cc7d27ecebaed78be","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-24 19:14:16.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"88b7731f620c7d035aad3a296301442d9babd970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:24:06.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78d83ba452bb4435992226790dd3e01d : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07b27913d884984aa985881b46eeda3 : SUCCESS in 4m 33s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":19},{"id":"dd7d9dfebbace40db422a4b4dc856c1ff5505bdd","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2022-01-25 04:32:58.000000000","message":"Patch Set 19:\n\n(7 comments)","accounts_in_message":[],"_revision_number":19},{"id":"c43f2af9545b3dcffb9b8c65f2bd962e0446ffee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 04:37:57.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b7dd60460414b22a24cab3562f53b13 : FAILURE in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9347e075009b4015a9029145ffe65f50 : SUCCESS in 3m 48s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":19},{"id":"83a0849a6511089ae7bec2c5fabd497c486b8023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-25 10:33:10.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"23bddfd381b262e907d0b9da165933716a9ed932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:38:10.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037495e6bcae423bbf777f6b7ad7b62a : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56180aa3dc0e49ca872411f1cface7e8 : FAILURE in 3m 18s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":20},{"id":"c9e23a18418644ee982d7ce05ef63aed61f3fa83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-25 12:01:58.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"ec6cd2e0230cabb28a7d96e47f4a99a5d4b88a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 12:09:22.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007fe7bee8b747eaac8ebf3e93e37ad3 : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363ad3bde1b24b589da915c2eea33e06 : SUCCESS in 3m 34s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/newton/linux-ficon-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/newton/linux-ficon-support.rst","accounts_in_message":[],"_revision_number":21},{"id":"0c61aeefdbabf1f7864f428dad1726855d25c159","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-01-25 15:05:18.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"71ef25913087b26c9b6d85206258f20c2746e890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-25 15:11:39.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"256c970a0333a424c71815fefe3b93b672871060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 15:20:14.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a754677516424f982e015c3a2b7eab : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374bf620d8c74f85aa8f76570102bbfd : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":22},{"id":"3b4540a86a3d7122c28623a10a0df1b8fdfd91d0","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-01-25 15:33:30.000000000","message":"Patch Set 22: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"e401795cf271e9a8416cb492b90440fa45f0a735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-25 15:57:35.000000000","message":"Patch Set 23: Commit message was updated.","accounts_in_message":[],"_revision_number":23},{"id":"d5811b482fb6d909d57a033efb2e0ae9261ae5c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-25 15:58:22.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"fa9ca943f2e35967780df1fdb5dffed95f1d792e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2022-01-25 15:58:59.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"a2cea64a13bf5da712d0a05a7eeacf0c94f01ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:13:04.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2805b7754c244dea9cd6b21e945ee79 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55566cb26f74cd09831559e079705ca : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":25},{"id":"620457cdb739efa11b42f3e603a0b7c59082117d","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-01-25 16:19:05.000000000","message":"Patch Set 25: -Code-Review","accounts_in_message":[],"_revision_number":25},{"id":"64b9ab67ebaebfa8f88b95d6d53cdfe2be2fe485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-01-25 16:51:23.000000000","message":"Patch Set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"8515113675ae88db97fea7dc357a606965bdae6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 17:02:22.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05c93194e6847db9f4caff845609ece : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de6390625a0e463eb990dd0865f5cd5e : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":26},{"id":"28b861d444208f81a48867fb28ef19bf5ca816e6","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2022-02-02 04:10:09.000000000","message":"Patch Set 26:\n\n(5 comments)","accounts_in_message":[],"_revision_number":26},{"id":"381d31da5b882c1ce8371831a29611ed8fc1bbee","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2022-02-02 04:10:48.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"2db6ca875d03fa441464b2500c36a031e42f8a11","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-03-03 17:37:52.000000000","message":"Patch Set 26: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"8b23fc44883ea25db0decfddc4cf6c837188fcba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-06-20 19:09:01.000000000","message":"Abandoned\n\nHas no response since 4 months and we\u0027re in the spec freeze week of Zed cycle. Feel free to restore if you would like to continue working on it.","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"65302b319d05bf9ea94a2fb2955c06b5c309b9d1","revisions":{"ef8bf2a63ab00a9a7751f253f4e5d1ff28686b18":{"kind":"REWORK","_number":1,"created":"2021-12-10 08:25:04.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/1"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ef8bf2a63ab00a9a7751f253f4e5d1ff28686b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ef8bf2a63ab00a9a7751f253f4e5d1ff28686b18"}]},"branch":"refs/heads/master"},"18a78201d9d6aba633794b697b1a335150327622":{"kind":"REWORK","_number":2,"created":"2021-12-10 10:21:30.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/2"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 10:21:30.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/18a78201d9d6aba633794b697b1a335150327622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/18a78201d9d6aba633794b697b1a335150327622"}]},"branch":"refs/heads/master"},"68fef62511ff7da1fb87241ffd8bac1064111ccc":{"kind":"REWORK","_number":3,"created":"2021-12-10 11:10:10.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/3"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 11:10:02.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/68fef62511ff7da1fb87241ffd8bac1064111ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/68fef62511ff7da1fb87241ffd8bac1064111ccc"}]},"branch":"refs/heads/master"},"fbd2ef2b504ea9f4782ec30557bc38cb14ee717f":{"kind":"REWORK","_number":4,"created":"2021-12-10 11:25:26.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/4"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 11:25:26.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/fbd2ef2b504ea9f4782ec30557bc38cb14ee717f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/fbd2ef2b504ea9f4782ec30557bc38cb14ee717f"}]},"branch":"refs/heads/master"},"d5296c4201095466168ed4d77586b79e2ebc605f":{"kind":"REWORK","_number":5,"created":"2021-12-10 11:49:02.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/5"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 11:49:01.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d5296c4201095466168ed4d77586b79e2ebc605f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d5296c4201095466168ed4d77586b79e2ebc605f"}]},"branch":"refs/heads/master"},"44be4be734b0b7c7b89e5d8863e9c256b436cc77":{"kind":"REWORK","_number":6,"created":"2021-12-13 11:38:03.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/6"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman","email":"Suman.Shaw1@ibm.com","date":"2021-12-13 11:38:03.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/44be4be734b0b7c7b89e5d8863e9c256b436cc77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/44be4be734b0b7c7b89e5d8863e9c256b436cc77"}]},"branch":"refs/heads/master"},"f30a0227ecd2058b2bc915e7fc586877d8977ccd":{"kind":"REWORK","_number":7,"created":"2021-12-13 11:39:25.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/7"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman","email":"Suman.Shaw1@ibm.com","date":"2021-12-13 11:39:14.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f30a0227ecd2058b2bc915e7fc586877d8977ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/f30a0227ecd2058b2bc915e7fc586877d8977ccd"}]},"branch":"refs/heads/master"},"916e01e937cf5b224783e5e317c5c4a2833bf2b2":{"kind":"REWORK","_number":8,"created":"2021-12-16 05:45:40.000000000","uploader":{"_account_id":32345,"name":"Suman","email":"Suman.Shaw1@ibm.com","username":"sumanshaw1"},"ref":"refs/changes/72/821372/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/8"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Suman","email":"Suman.Shaw1@ibm.com","date":"2021-12-16 05:45:35.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/916e01e937cf5b224783e5e317c5c4a2833bf2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/916e01e937cf5b224783e5e317c5c4a2833bf2b2"}]},"branch":"refs/heads/master"},"af6f958e583f244ecfb4c5ab89327580cf33fb3b":{"kind":"REWORK","_number":9,"created":"2022-01-16 13:55:52.000000000","uploader":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"ref":"refs/changes/72/821372/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/9"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","date":"2022-01-16 13:55:34.000000000","tz":-300},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/af6f958e583f244ecfb4c5ab89327580cf33fb3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/af6f958e583f244ecfb4c5ab89327580cf33fb3b"}]},"branch":"refs/heads/master"},"757dc477e0d1c50e202b6a77e2d801c831cad746":{"kind":"REWORK","_number":10,"created":"2022-01-23 17:54:55.000000000","uploader":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"ref":"refs/changes/72/821372/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/10"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"chhagarw","email":"chhagarw@in.ibm.com","date":"2022-01-23 17:54:29.000000000","tz":-300},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/757dc477e0d1c50e202b6a77e2d801c831cad746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/757dc477e0d1c50e202b6a77e2d801c831cad746"}]},"branch":"refs/heads/master"},"1f07f8c4da2b24f52429b5095bc2d2d77cc4b891":{"kind":"REWORK","_number":11,"created":"2022-01-24 08:47:02.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/11"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 08:47:01.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1f07f8c4da2b24f52429b5095bc2d2d77cc4b891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1f07f8c4da2b24f52429b5095bc2d2d77cc4b891"}]},"branch":"refs/heads/master"},"8f2993b52451a1398704a0c997f2ca0d6fc6fbd8":{"kind":"REWORK","_number":12,"created":"2022-01-24 11:03:46.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/12"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 11:03:41.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8f2993b52451a1398704a0c997f2ca0d6fc6fbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8f2993b52451a1398704a0c997f2ca0d6fc6fbd8"}]},"branch":"refs/heads/master"},"42ad14e7549249b8b88011462d11537609582d96":{"kind":"REWORK","_number":13,"created":"2022-01-24 11:23:52.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/13"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 11:23:47.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/42ad14e7549249b8b88011462d11537609582d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/42ad14e7549249b8b88011462d11537609582d96"}]},"branch":"refs/heads/master"},"a901b0afbbda592f838ca11b32c4333fd212338d":{"kind":"REWORK","_number":14,"created":"2022-01-24 11:39:44.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/14"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 11:39:40.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a901b0afbbda592f838ca11b32c4333fd212338d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a901b0afbbda592f838ca11b32c4333fd212338d"}]},"branch":"refs/heads/master"},"b4585bd2d2139b7624aa914abe6f72dfbb36baf2":{"kind":"REWORK","_number":15,"created":"2022-01-24 11:57:17.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/15"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 11:57:17.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b4585bd2d2139b7624aa914abe6f72dfbb36baf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b4585bd2d2139b7624aa914abe6f72dfbb36baf2"}]},"branch":"refs/heads/master"},"c3d7f2200a04838c6b2d3c2c650a4e36cb9c6ae4":{"kind":"REWORK","_number":16,"created":"2022-01-24 14:09:25.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/16"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 14:09:24.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c3d7f2200a04838c6b2d3c2c650a4e36cb9c6ae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c3d7f2200a04838c6b2d3c2c650a4e36cb9c6ae4"}]},"branch":"refs/heads/master"},"2d4f11b92ba47d2cb41c16be00e065f7060124bb":{"kind":"REWORK","_number":17,"created":"2022-01-24 14:34:44.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/17"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 14:34:44.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2d4f11b92ba47d2cb41c16be00e065f7060124bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2d4f11b92ba47d2cb41c16be00e065f7060124bb"}]},"branch":"refs/heads/master"},"56875fdc6b60cd32d22237b816d0316458f423bf":{"kind":"REWORK","_number":18,"created":"2022-01-24 15:33:19.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/18"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 15:33:06.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/56875fdc6b60cd32d22237b816d0316458f423bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/56875fdc6b60cd32d22237b816d0316458f423bf"}]},"branch":"refs/heads/master"},"76621c1853789a6eb44e594eae15ce2c85a4a4d1":{"kind":"REWORK","_number":19,"created":"2022-01-24 15:37:38.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/19"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-24 15:37:31.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/76621c1853789a6eb44e594eae15ce2c85a4a4d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/76621c1853789a6eb44e594eae15ce2c85a4a4d1"}]},"branch":"refs/heads/master"},"353c751f3cb288720010713917a558c92d523b1e":{"kind":"REWORK","_number":20,"created":"2022-01-25 10:33:10.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/20"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-25 10:33:10.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/353c751f3cb288720010713917a558c92d523b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/353c751f3cb288720010713917a558c92d523b1e"}]},"branch":"refs/heads/master"},"c359290daeda2ac325e1746318be2c2939bf5328":{"kind":"REWORK","_number":21,"created":"2022-01-25 12:01:58.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/21"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-25 12:01:58.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c359290daeda2ac325e1746318be2c2939bf5328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c359290daeda2ac325e1746318be2c2939bf5328"}]},"branch":"refs/heads/master"},"207dce4b20cc7639f377e248ab7299e3bcee5396":{"kind":"REWORK","_number":22,"created":"2022-01-25 15:11:39.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/22"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-25 15:11:39.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/207dce4b20cc7639f377e248ab7299e3bcee5396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/207dce4b20cc7639f377e248ab7299e3bcee5396"}]},"branch":"refs/heads/master"},"8e59cb01cc02291a9c8865641cac889de309a9e3":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2022-01-25 15:57:35.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/23"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-25 15:57:35.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nDepends-on: https://review.opendev.org/c/openstack/cinder-specs/+/826302/\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8e59cb01cc02291a9c8865641cac889de309a9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8e59cb01cc02291a9c8865641cac889de309a9e3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"299bf5b12b7b047bfbef4085f5ca353f9b4ab80a":{"kind":"REWORK","_number":24,"created":"2022-01-25 15:58:22.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/24"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-25 15:58:12.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nDepends-on: https://review.opendev.org/c/openstack/cinder-specs/+/826302/\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/299bf5b12b7b047bfbef4085f5ca353f9b4ab80a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/299bf5b12b7b047bfbef4085f5ca353f9b4ab80a"}]},"branch":"refs/heads/master"},"077a9b60405f7249d1feb9e5b4d2c46b3dedc6a8":{"kind":"REWORK","_number":25,"created":"2022-01-25 15:58:59.000000000","uploader":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"ref":"refs/changes/72/821372/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/25"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"prashkre","email":"prashkre@in.ibm.com","date":"2022-01-25 15:58:50.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nDepends-on: https://review.opendev.org/c/openstack/cinder-specs/+/826302/\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/077a9b60405f7249d1feb9e5b4d2c46b3dedc6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/077a9b60405f7249d1feb9e5b4d2c46b3dedc6a8"}]},"branch":"refs/heads/master"},"65302b319d05bf9ea94a2fb2955c06b5c309b9d1":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2022-01-25 16:51:23.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/72/821372/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/72/821372/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/72/821372/26"}}},"commit":{"parents":[{"commit":"7464b781cfcfcd334be97b20b3f66fc9518f7f9b","subject":"Add yoga directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7464b781cfcfcd334be97b20b3f66fc9518f7f9b"}]}],"author":{"name":"Suman Shaw","email":"Suman.Shaw1@ibm.com","date":"2021-12-10 08:24:25.000000000","tz":330},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2022-01-25 16:51:23.000000000","tz":0},"subject":"Add multiple volume groups - Yoga","message":"Add multiple volume groups - Yoga\n\nAdd multiple volume groups for yoga\n\nChange-Id: I9fd0700123813b1f7afc4e86177c4c67f43caaae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/65302b319d05bf9ea94a2fb2955c06b5c309b9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/65302b319d05bf9ea94a2fb2955c06b5c309b9d1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
