)]}'
{"id":"openstack%2Fcinder-specs~864020","triplet_id":"openstack%2Fcinder-specs~master~If7168afad1a0e53b9642478023bd4c20801f1916","project":"openstack/cinder-specs","branch":"master","topic":"bp/use-assisted-volume-extend-api","attention_set":{},"removed_from_attention_set":{"34859":{"account":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"last_update":"2022-12-19 11:34:03.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-12-05 04:08:55.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"If7168afad1a0e53b9642478023bd4c20801f1916","subject":"Use assisted volume extend API","status":"ABANDONED","created":"2022-11-08 15:46:47.000000000","updated":"2022-12-19 11:34:03.000000000","total_comment_count":25,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"707fc61ea33992b8569482e2d7a1188b84fa0179","_number":864020,"virtual_id_number":864020,"owner":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-23 14:13:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-05 04:08:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 15:59:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-16 08:49:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"25e2c40038218b10b65112a230f96eef1e698248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-11-08 15:46:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb6e2db556974951042d1af76da84ce62ecf12d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 15:59:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240fc19c41d94c8b9306d337ba1dfad3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12f91f5fca04fffb45d8687ea65d6fb : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f728d363c4450d9dd4453229c96d85 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"575cfc02140280977fa2e960e22810c0881a5148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-11-09 10:33:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"caacc6bea6df85442fe518c692f3ec1ab50d65a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:37:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae3127c144a24031bc9bffdb86f69815\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edcc5e1724a4ebca481520bfb7efad5 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/234398865bf147c8a022bb2e3ee9d3f6 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"fa7d37e012dba8ad849953b104eacc11bb018b28","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-16 08:49:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3f78eb28633abb54d74651a9b535a3a6c36834e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-11-23 14:07:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"45d36f535efffa849e49061fa28662501c0addb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 14:13:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ec3e5526a347cba1b81e4eca67a818\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54e469ffc4e48409f2e6518d790c34f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e71737053a64085a8888ca1681b75e8 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d8012d08e44b7c0022e34af440a7cc86541d7e04","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-11-23 14:57:17.000000000","message":"Patch Set 3:\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8c2272a5100c1ac38a8c46147faa5850324e7973","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-05 04:08:55.000000000","message":"Patch Set 3: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"707fc61ea33992b8569482e2d7a1188b84fa0179","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-19 11:34:03.000000000","message":"Abandoned\n\nReplaced by https://review.opendev.org/c/openstack/cinder-specs/+/866718","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b36f6e0f13d65452553222487a4a12a6789dc7f","revisions":{"b4f6604f82b5de649e9a79862f6dd099aa0340cf":{"kind":"REWORK","_number":1,"created":"2022-11-08 15:46:47.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/20/864020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/20/864020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/1"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-11-08 15:43:14.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-11-08 15:46:45.000000000","tz":60},"subject":"Use assisted volume extend API","message":"Use assisted volume extend API\n\nThis blueprint proposes to use the assisted volume extend API that has\nbeen proposed for Nova, to support online extend for remotefs-based volume\ndrivers such as NFS, NetApp NFS, and Quobyte.\n\nChange-Id: If7168afad1a0e53b9642478023bd4c20801f1916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b4f6604f82b5de649e9a79862f6dd099aa0340cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b4f6604f82b5de649e9a79862f6dd099aa0340cf"}]},"branch":"refs/heads/master"},"075b4e2c14d7b00bb1b2be49a2d06cf35df2998e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-09 10:33:08.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/20/864020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/20/864020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/2"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-11-08 15:43:14.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-11-09 10:27:53.000000000","tz":60},"subject":"Use assisted volume extend API","message":"Use assisted volume extend API\n\nThis spec proposes to use the assisted volume extend API that has been\nproposed for Nova, to support online extend for remotefs-based volume\ndrivers such as NFS, NetApp NFS, and Quobyte.\n\nBlueprint: use-assisted-volume-extend-api\nChange-Id: If7168afad1a0e53b9642478023bd4c20801f1916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/075b4e2c14d7b00bb1b2be49a2d06cf35df2998e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/075b4e2c14d7b00bb1b2be49a2d06cf35df2998e"}]},"branch":"refs/heads/master"},"6b36f6e0f13d65452553222487a4a12a6789dc7f":{"kind":"REWORK","_number":3,"created":"2022-11-23 14:07:17.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/20/864020/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/20/864020/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/20/864020/3"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-11-08 15:43:14.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-11-23 14:06:03.000000000","tz":60},"subject":"Use assisted volume extend API","message":"Use assisted volume extend API\n\nThis spec proposes to use the assisted volume extend API that has been\nproposed for Nova, to support online extend for remotefs-based volume\ndrivers such as NFS, NetApp NFS, Powerstore NFS, and Quobyte.\n\nBlueprint: use-assisted-volume-extend-api\nChange-Id: If7168afad1a0e53b9642478023bd4c20801f1916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6b36f6e0f13d65452553222487a4a12a6789dc7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6b36f6e0f13d65452553222487a4a12a6789dc7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
