)]}'
{"id":"openstack%2Fcinder-specs~866718","triplet_id":"openstack%2Fcinder-specs~master~Iffe5e0e05287d57b27227d66864ee226424b5cd4","project":"openstack/cinder-specs","branch":"master","topic":"bp/extend-volume-completion-action","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-12-24 17:48:48.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-12-24 17:48:48.000000000","reason":"Change was submitted"},"34859":{"account":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"last_update":"2022-12-24 17:48:48.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-12-24 17:48:48.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-12-24 17:48:48.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-12-24 17:40:04.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Iffe5e0e05287d57b27227d66864ee226424b5cd4","subject":"Add extend volume completion action","status":"MERGED","created":"2022-12-06 14:11:12.000000000","updated":"2023-01-09 17:40:32.000000000","submitted":"2022-12-24 17:48:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":68,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"866718-bp/extend-volume-completion-action","meta_rev_id":"05c3c38b4fc71fe4c274c77f948827c545d91dea","_number":866718,"virtual_id_number":866718,"owner":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-24 17:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-24 17:40:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-24 17:40:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 14:19:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-14 14:29:36.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-12-14 14:35:02.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-12-15 14:16:36.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-12-21 13:58:50.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-12-23 20:41:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"d007ac4b8b35d35661a0c19d2b99366b142beffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-06 14:11:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d74c4002732149457fd276f34a1291d876b528b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:19:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be6f1543aac64501826a819e39cd27ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea0798416f4441c8f8119b400585d89 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f8831525904d068bd53454f286ef3b : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3bf9cb7ae81ec068c2032d35339d5cb2a971d486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-06 14:33:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7ada37012e6b58c17d92794ea7112a77fb35896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:38:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b5ff3257fb4e0e9b5d258369d5251d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7b6a3a4a984421a1eca7087c6808d1 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a304b791244e45aea54bb87baad1dd : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f35298b07c3513cefeafbc42c2eb1653c8b52483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-08 11:59:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"20f04541d7cc61746b813e2d813a1413f24e0a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 12:05:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2ae850814da41e7945b263c80f11d06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3193a9edbe047eb9c4be421d742f96d : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3b2768a59d466b9191d35b33b3a2c2 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7e2307e5fc3ff08de01d6e98229d52f0e44abcab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-08 14:11:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6e3444f97774fcc6689c5b41a09f3cdb0e9a41b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 14:21:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b5a54988df40428413143ce16bd02f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e44ee1f070843328729d807a24505e1 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af14540e3d84a88a176c8226a034040 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":4},{"id":"11931427fbad39ad7359a93dbff9efd6298201eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-13 17:46:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4f835fa4e2730417fdf1d8c5a2d9d7e6e96604a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 17:58:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93b0cbbf50de4a968a346114da7114c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ad50665a444be9b297d579f8bbe977 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33bcc8c685d4091824e64012390a1fe : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":5},{"id":"ee3d8873314215e589ded71f5336dbc1f41c3bc7","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-14 09:16:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ba20046b79d966528e0930d038ff149a3dcc4e00","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-12-14 14:35:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"80ca971e7c44b20e0bc8738da08ca33241dd263a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-14 14:49:27.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"5e8cbd825dafd9be7dc45d2170cb4d06c977e44d","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-14 14:50:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6a9eaafa0139945c32da8baf7db0c6fa7a428a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 14:54:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26bfe7c3a09b41ed8474f8094e2e9678\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7bf6a103e3c46b3a35ea68dd6f3862e : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/231dee00cb334f61809d111b82689e73 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":6},{"id":"206b6e7f030c162d173e4f54a7ad42b323c4c7b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-15 11:01:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"665e98d03aa25281259338c253819e8cd290c5b4","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-15 11:06:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a4dba10df3681e0f44fc8eaa7a7205bf34bf76aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 11:11:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d18ef9cf2b814ecaba5f234a93921dd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a386a7775b544bc9d79f41239a97248 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4fb58b0f8b4e2eafea0820e4310477 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":7},{"id":"64fc65ac43ff941d28fa0144fd0e7937ee374010","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-12-15 14:16:36.000000000","message":"Patch Set 7: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9e06e69befff9e138f1df6c69330f9f9a2d9a8c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-19 15:04:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cac359de3a02bf6674e1b93999edb3c2d5c92ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:09:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78faf74fadcc4905bc1bb7bdfef57ad0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe67b66cdb3487ca1ade905c496c21e : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0798c33ec9d0449080d05b4ba4a16f07 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":8},{"id":"ea6d84f09a3bbba0429a1cd5ff89c77295e96025","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-19 15:21:30.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"94403f7ee929bee42c0149c722ca2d080b6ac3a6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-21 13:58:50.000000000","message":"Patch Set 8: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d929f9e6926cbf2fbb1e07ab3a439c219df7dcf0","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-21 16:06:04.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3a9ddf01d2d07a06672ab2f6ff5eabe15c683f22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-22 13:19:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"00f44fa4a3a907efa817a77465c492c088d51769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 13:25:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7abfee50d3e3430eb79ffd7f6894eabf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afd87cff7ba41fd8dd5051280206fee : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d3ff29412c4c2ea87fc81b4c7c288e : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":9},{"id":"6cff1a5153cf40a46a82bf07d9bc30f6ce727627","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-22 13:33:58.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c6ea1ae51c513bba0a7e85b1b5e12518f82d0502","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-22 15:55:50.000000000","message":"Patch Set 9: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5c766014f775a07ebe6fde9a5838e9e8a1f7c4d4","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-22 19:19:57.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"72c03ef335e643131b2a5c14cf95274f72e025d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-22 20:30:22.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"09dab169e1cc1ad1cf30783006246346a499a007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 20:45:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b1bb0b9b562458d8bcf2ef97824eefa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76441da0dbcb4d9b8687d2e1611d278d : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16dd37a8dfa44ba2a668b16864d5b992 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":10},{"id":"8d7c84ad76dcd685e21e40b0d0b5652148c97285","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-23 11:15:44.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"690a21b179c821fadba3ad58845b998685295cae","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-23 13:15:21.000000000","message":"Patch Set 10: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b989538f49f99db880ee8a480b6e91c5f48e59dc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-23 20:41:12.000000000","message":"Patch Set 10: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a4401f5997942822b3061577d30738e3bbcb95ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-24 10:53:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"07b49739d2dccb4c061a674e4d08105c6905c20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 10:58:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc177e7b7614651953e2ce50d4a364a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fac3fa6911b4648a19f1b85080dd459 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6e2c69d04c42f7bc0cd56a01a9116e : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":11},{"id":"004291a38ef8c8abac31489f42becce8ebd755bc","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2022-12-24 11:00:41.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"5e6d719a8282b9d5d1079f635a164762c60b2794","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-24 17:40:04.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"30a9a20de6f6f99d3f126ec00ab3abe6d2fe1bdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 17:40:25.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"0e840204fe10f793a0dd94ef1c70d5ff0bb96849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 17:48:48.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b295efe2ad8248849d98a1d64e3ccafb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67ed666f4e2c49b8a71c2aa5e1294409 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d803393fed42b2942f17db3cfb663c : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":11},{"id":"c2dc608f721980e5197548fe703b22aa01bcb2bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 17:48:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"83a8ac36e1068e167206a3eb1fe63dcf993d58c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 17:49:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41396ed5dc904debb4c2724da535cd7c\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/b05379404d634c45b2ed3a9024dee440 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":11},{"id":"5d483eb3c3aa9c30687dfaf510c7550c26f3e17a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-03 18:14:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"31fa4e006e2e08dc6316271f6e2efc7056d1679b","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-01-04 09:52:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"05c3c38b4fc71fe4c274c77f948827c545d91dea","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-09 17:40:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"52cf89058464e4d2f91286b7590d6aa493929deb","revisions":{"d8df4e6410452476360b9ee6b59d9ed4c918dacb":{"kind":"REWORK","_number":1,"created":"2022-12-06 14:11:12.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/1"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:11:11.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes to use the assisted volume extend API that has been\nThis blueprint proposes a new volume action that can be used by Nova\nto notify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d8df4e6410452476360b9ee6b59d9ed4c918dacb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d8df4e6410452476360b9ee6b59d9ed4c918dacb"}]},"branch":"refs/heads/master"},"4244879a39ba976c2cccb2fa5b5dd57b141b729c":{"kind":"REWORK","_number":2,"created":"2022-12-06 14:33:13.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/2"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:32:28.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes to use the assisted volume extend API that has been\nThis blueprint proposes a new volume action that can be used by Nova\nto notify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4244879a39ba976c2cccb2fa5b5dd57b141b729c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/4244879a39ba976c2cccb2fa5b5dd57b141b729c"}]},"branch":"refs/heads/master"},"8a0dd017083eb70bbacc017299946a7fb2f07271":{"kind":"REWORK","_number":3,"created":"2022-12-08 11:59:14.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/3"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-08 11:59:01.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes to use the assisted volume extend API that has been\nThis blueprint proposes a new volume action that can be used by Nova\nto notify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8a0dd017083eb70bbacc017299946a7fb2f07271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/8a0dd017083eb70bbacc017299946a7fb2f07271"}]},"branch":"refs/heads/master"},"648df134269bf722f2046f907ee0a4067e91c4ab":{"kind":"REWORK","_number":4,"created":"2022-12-08 14:11:05.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/4"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-08 14:10:47.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes to use the assisted volume extend API that has been\nThis blueprint proposes a new volume action that can be used by Nova\nto notify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/648df134269bf722f2046f907ee0a4067e91c4ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/648df134269bf722f2046f907ee0a4067e91c4ab"}]},"branch":"refs/heads/master"},"90d1891e135bc96bfabfb83d20b4d0b5f1e2d3ab":{"kind":"REWORK","_number":5,"created":"2022-12-13 17:46:12.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/5"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-13 17:45:56.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes to use the assisted volume extend API that has been\nThis blueprint proposes a new volume action that can be used by Nova\nto notify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/90d1891e135bc96bfabfb83d20b4d0b5f1e2d3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/90d1891e135bc96bfabfb83d20b4d0b5f1e2d3ab"}]},"branch":"refs/heads/master"},"ec48d8cd5416fce5a5f3bd409022bdd6b241bee9":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-12-14 14:49:27.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/6"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-14 14:47:44.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes a new volume action that can be used by Nova to\nnotify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ec48d8cd5416fce5a5f3bd409022bdd6b241bee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ec48d8cd5416fce5a5f3bd409022bdd6b241bee9"}]},"branch":"refs/heads/master"},"3cb14db451d82c72e5abf2c69c35b351901dc0df":{"kind":"REWORK","_number":7,"created":"2022-12-15 11:01:24.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/7"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-15 11:01:13.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes a new volume action that can be used by Nova to\nnotify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3cb14db451d82c72e5abf2c69c35b351901dc0df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3cb14db451d82c72e5abf2c69c35b351901dc0df"}]},"branch":"refs/heads/master"},"a354d3643190f7f94e2d040efc4476d7b82c4c05":{"kind":"REWORK","_number":8,"created":"2022-12-19 15:04:20.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/8"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-19 15:04:08.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes a new volume action that can be used by Nova to\nnotify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a354d3643190f7f94e2d040efc4476d7b82c4c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/a354d3643190f7f94e2d040efc4476d7b82c4c05"}]},"branch":"refs/heads/master"},"5699cc6df51aecb93de50e9f3d15fac4889af896":{"kind":"REWORK","_number":9,"created":"2022-12-22 13:19:42.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/9"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-22 13:19:40.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes a new volume action that can be used by Nova to\nnotify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5699cc6df51aecb93de50e9f3d15fac4889af896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/5699cc6df51aecb93de50e9f3d15fac4889af896"}]},"branch":"refs/heads/master"},"26d9c6bac09fb5535dd28fa1aa7506b6206d9b6a":{"kind":"REWORK","_number":10,"created":"2022-12-22 20:30:22.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/10"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-22 20:29:38.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes a new volume action that can be used by Nova to\nnotify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/26d9c6bac09fb5535dd28fa1aa7506b6206d9b6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/26d9c6bac09fb5535dd28fa1aa7506b6206d9b6a"}]},"branch":"refs/heads/master"},"52cf89058464e4d2f91286b7590d6aa493929deb":{"kind":"REWORK","_number":11,"created":"2022-12-24 10:53:13.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/18/866718/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/18/866718/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/18/866718/11"}}},"commit":{"parents":[{"commit":"0c97a9b3f42516a03703e95805879cddae882e84","subject":"Merge \"Add redirect to zed for image encryption spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/0c97a9b3f42516a03703e95805879cddae882e84"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-06 14:05:18.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2022-12-24 10:52:58.000000000","tz":60},"subject":"Add extend volume completion action","message":"Add extend volume completion action\n\nThis spec proposes a new volume action that can be used by Nova to\nnotify Cinder on success or failure when handling \"volume-extended\"\nexternal server events.\nThe new volume action is used add support for extending attached volumes\nto the NFS, NetApp NFS, Powerstore NFS, and Quobyte volume drivers.\n\nBlueprint: extend-volume-completion-action\nChange-Id: Iffe5e0e05287d57b27227d66864ee226424b5cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52cf89058464e4d2f91286b7590d6aa493929deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52cf89058464e4d2f91286b7590d6aa493929deb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}]}],"submit_requirements":[]}
