)]}'
{"id":"openstack%2Fcinder-specs~872019","triplet_id":"openstack%2Fcinder-specs~master~I5f25aa2eca758a00a7402d1f3b89610aa715eea4","project":"openstack/cinder-specs","branch":"master","topic":"bp/allocated-size-distinction","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2023-06-13 11:59:54.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-06-13 11:59:54.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2023-11-22 14:22:40.000000000","reason":"Change was abandoned"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-06-13 11:59:54.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-06-13 11:58:07.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I5f25aa2eca758a00a7402d1f3b89610aa715eea4","subject":"Sizing capacity rounded volumes","status":"ABANDONED","created":"2023-01-27 14:09:31.000000000","updated":"2023-11-22 14:22:40.000000000","total_comment_count":38,"unresolved_comment_count":21,"has_review_started":true,"meta_rev_id":"89fc711ac0fc276b2c8a2e1eb889c758ee020d09","_number":872019,"virtual_id_number":872019,"owner":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-06-12 14:08:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-06-13 11:58:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-06-14 14:16:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 14:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 15:06:41.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-01-27 15:06:41.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-02-09 14:28:38.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2023-03-21 18:59:59.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-06-13 11:59:54.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"}],"messages":[{"id":"77e5638c39994996d2e60e8989dbd1fad0367aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-27 14:09:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdc7ac838c2cd6d903caf03134f71018f7c041ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 14:18:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/897fa92b2e5a4c879638977e08eb2375\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef5847198f44deb87618c89136625e7 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3257137efc64f109468ded58428901e : FAILURE in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6b92e986511e26537276c828fa31e1ff0232e838","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-27 14:25:44.000000000","message":"Abandoned\n\nwrong directory","accounts_in_message":[],"_revision_number":1},{"id":"0af6b6e7c49727a45116d6be24247b0eb9de64ed","tag":"autogenerated:gerrit:restore","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-27 14:53:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"f1b3634cd1dd4d0ffaacd594e8a69de0fdf79d5b","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-27 14:54:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2679a536cf57c71abfc6e6a100a21a4ddfe832e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-27 14:56:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac9519d7c454c6f148f5d937e774009263ccf838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:03:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1869cd41fd0f46408504389ded06ff53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16949498d30f42999433f139fbf773bc : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a155affccd4dd8aa5b3755a8ffb9b3 : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"593cea2a5c4a108a91c05a1503aaaea16d0e0ef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-27 15:04:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3b714b85661badeeef6851a0e77f8660eb16bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:10:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf77646b26c4cd88feec023766d784b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce78a0f35a845ebb6165c457156aa13 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7507088896c49d88d3c14e6658d8815 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"7b609721b6f097f59cab10ac1436f6c3dbf6d47c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-27 15:12:07.000000000","message":"Topic set to bp/allocated-size-distinction","accounts_in_message":[],"_revision_number":3},{"id":"c58cb8dc2ba554b158d736d1a07897552368a4c4","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-02-09 14:28:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b73c4c8cfda57de1024c0b254978d4b29e743067","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-02-09 14:32:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0b781410a32b8870cb8adc33fad48104310d3759","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-02-09 14:42:07.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b7151c0cabc5152e9cbffaaaeedccbcf185febb","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-02-09 14:44:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"04eeacba9e4eeab511522452eaeb3a28ddc5e240","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-21 18:59:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81cab697acc5bb157e8ff97b6cfbf7cd9968c391","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-04-26 17:09:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3978a8665b1389ed0f130e7c44944a025cfb9c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-04-27 13:25:28.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e1e3f33997f553a2680435fd83b59e60661fc324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 13:30:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18535c6fd4944bdabe3865157ff1a3ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3072819fbf604bb798296618d8d3b15a : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cff5dbaa686401b93de8ffd6972127f : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":4},{"id":"3663e67e1fa09df77cb86b2083ecc99bb69c9ca4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-06-09 20:27:01.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"35e109ea584863c552e6974024ab02b43b89a8cd","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-06-12 14:01:43.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"372a5cc57c3dfc550b9237b981a67f7148e69635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-06-12 14:02:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c26012f204ad0d90a7786fc08d3e35bdce1c45d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 14:08:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26a926c9a69043a38e30f0ab19cff289\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5cf44407bff4556915cb4dda7b0fc97 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2ddf38147247a4be0e797f72f079b0 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":5},{"id":"df4e350da9586507b94b02cf273af8f3c991ce15","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-13 11:58:07.000000000","message":"Patch Set 5: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8bc9dc6e343dd386337a6a835127260f745148df","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-13 11:59:54.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ea7011683e04d2b9e9082d693baa5e390bee95c0","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-14 14:16:10.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2b257e562f0f5c48e9dc01b8ed23418dd183e760","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-11-22 12:03:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"89fc711ac0fc276b2c8a2e1eb889c758ee020d09","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-11-22 14:22:40.000000000","message":"Abandoned\n\nNot enough bandwidth and ask for this currently, can be restored or reproposed if needed.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9ce0b6c99b9bd122ad3b412627a2fa1d2c8b3f14","revisions":{"facb9a4d34b5227007b6371e5393393ae9739cb7":{"kind":"REWORK","_number":1,"created":"2023-01-27 14:09:31.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/19/872019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/19/872019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/1"}}},"commit":{"parents":[{"commit":"52cf89058464e4d2f91286b7590d6aa493929deb","subject":"Add extend volume completion action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52cf89058464e4d2f91286b7590d6aa493929deb"}]}],"author":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 14:04:28.000000000","tz":60},"committer":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 14:09:30.000000000","tz":60},"subject":"Sizing capacity rounded volumes","message":"Sizing capacity rounded volumes\n\nThis feature will support storage providers which use a rounding\nincrement to allocate space on the backend.\nThis will prevent the discrepency between what the end user has\nasked and what is being presented from the backend as a volume\n\nImplements: blueprint allocated-size-distinction\nChange-Id: I5f25aa2eca758a00a7402d1f3b89610aa715eea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/facb9a4d34b5227007b6371e5393393ae9739cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/facb9a4d34b5227007b6371e5393393ae9739cb7"}]},"branch":"refs/heads/master"},"1963171b878e0a0e089aec039afd95409bf92ddb":{"kind":"REWORK","_number":2,"created":"2023-01-27 14:56:57.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/19/872019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/19/872019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/2"}}},"commit":{"parents":[{"commit":"52cf89058464e4d2f91286b7590d6aa493929deb","subject":"Add extend volume completion action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52cf89058464e4d2f91286b7590d6aa493929deb"}]}],"author":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 14:04:28.000000000","tz":60},"committer":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 14:56:50.000000000","tz":60},"subject":"Sizing capacity rounded volumes","message":"Sizing capacity rounded volumes\n\nThis feature will support storage providers which use a rounding\nincrement to allocate space on the backend.\nThis will prevent the discrepency between what the end user has\nasked and what is being presented from the backend as a volume\n\nImplements: blueprint allocated-size-distinction\nChange-Id: I5f25aa2eca758a00a7402d1f3b89610aa715eea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1963171b878e0a0e089aec039afd95409bf92ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1963171b878e0a0e089aec039afd95409bf92ddb"}]},"branch":"refs/heads/master"},"d614688d8b1c3c24235d131015cf269bd3dd833d":{"kind":"REWORK","_number":3,"created":"2023-01-27 15:04:47.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/19/872019/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/19/872019/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/3"}}},"commit":{"parents":[{"commit":"52cf89058464e4d2f91286b7590d6aa493929deb","subject":"Add extend volume completion action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52cf89058464e4d2f91286b7590d6aa493929deb"}]}],"author":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 14:04:28.000000000","tz":60},"committer":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 15:04:34.000000000","tz":60},"subject":"Sizing capacity rounded volumes","message":"Sizing capacity rounded volumes\n\nThis feature will support storage providers which use a rounding\nincrement to allocate space on the backend.\nThis will prevent the discrepency between what the end user has\nasked and what is being presented from the backend as a volume\n\nImplements: blueprint allocated-size-distinction\nChange-Id: I5f25aa2eca758a00a7402d1f3b89610aa715eea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d614688d8b1c3c24235d131015cf269bd3dd833d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d614688d8b1c3c24235d131015cf269bd3dd833d"}]},"branch":"refs/heads/master"},"466087ef3fe4d8d7ef26a2b2abfda319604a1b04":{"kind":"REWORK","_number":4,"created":"2023-04-27 13:25:28.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/19/872019/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/19/872019/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/4"}}},"commit":{"parents":[{"commit":"52cf89058464e4d2f91286b7590d6aa493929deb","subject":"Add extend volume completion action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52cf89058464e4d2f91286b7590d6aa493929deb"}]}],"author":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 14:04:28.000000000","tz":60},"committer":{"name":"Jean Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-04-27 13:25:28.000000000","tz":0},"subject":"Sizing capacity rounded volumes","message":"Sizing capacity rounded volumes\n\nThis feature will support storage providers which use a rounding\nincrement to allocate space on the backend.\nThis will prevent the discrepency between what the end user has\nasked and what is being presented from the backend as a volume\n\nImplements: blueprint allocated-size-distinction\nChange-Id: I5f25aa2eca758a00a7402d1f3b89610aa715eea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/466087ef3fe4d8d7ef26a2b2abfda319604a1b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/466087ef3fe4d8d7ef26a2b2abfda319604a1b04"}]},"branch":"refs/heads/master"},"9ce0b6c99b9bd122ad3b412627a2fa1d2c8b3f14":{"kind":"REWORK","_number":5,"created":"2023-06-12 14:02:21.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/19/872019/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/19/872019/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/19/872019/5"}}},"commit":{"parents":[{"commit":"52cf89058464e4d2f91286b7590d6aa493929deb","subject":"Add extend volume completion action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52cf89058464e4d2f91286b7590d6aa493929deb"}]}],"author":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-27 14:04:28.000000000","tz":60},"committer":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-06-12 14:02:18.000000000","tz":120},"subject":"Sizing capacity rounded volumes","message":"Sizing capacity rounded volumes\n\nThis feature will support storage providers which use a rounding\nincrement to allocate space on the backend.\nThis will prevent the discrepency between what the end user has\nasked and what is being presented from the backend as a volume\n\nImplements: blueprint allocated-size-distinction\nChange-Id: I5f25aa2eca758a00a7402d1f3b89610aa715eea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9ce0b6c99b9bd122ad3b412627a2fa1d2c8b3f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/9ce0b6c99b9bd122ad3b412627a2fa1d2c8b3f14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
