)]}'
{"id":"openstack%2Fcinder-specs~877230","triplet_id":"openstack%2Fcinder-specs~master~I40d1632953cac767c66f735a11e0477afd63426f","project":"openstack/cinder-specs","branch":"master","topic":"bp/extend-volume-completion-action","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2023-04-14 11:06:25.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"34859":{"account":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"last_update":"2023-06-21 05:13:51.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-04-14 06:37:41.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I40d1632953cac767c66f735a11e0477afd63426f","subject":"Re-propose extend volume completion action","status":"MERGED","created":"2023-03-13 08:39:02.000000000","updated":"2023-06-21 05:14:49.000000000","submitted":"2023-06-21 05:13:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877230-bp/extend-volume-completion-action","meta_rev_id":"b284316a90d170855179b6f91b90fbf999f15bba","_number":877230,"virtual_id_number":877230,"owner":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 05:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":2,"date":"2023-04-26 13:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2023-04-14 11:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-09 12:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-21 05:09:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 08:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 14:59:05.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-04-14 06:37:41.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2023-04-14 11:06:25.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-04-26 13:55:40.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"}],"messages":[{"id":"58195d1bba2b0a93986b635d829a2a43e5034fee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-03-13 08:39:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"023b9f995d972df73a53c9c78afc7cd044bca66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 08:48:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c9258695c34b8086bd78309be6b6a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca683bac054a482190ff952859493a2c : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd9a20e582b434786e3c9c20c1a9cf7 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2f1ef4ce10b6f4788eaf7941ef57cb46e08af229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-04-04 08:29:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bf41e4c13261d953376b8d1c1ac5d178cb26286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 08:37:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9559868af4bc4ff88c4154dfba522a1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f39a77ace49409fad8a2446c399a00e : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77dbf031634145538ba5f2385e3688cc : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"61a33849ee1200bf5f3bd761e8358542a9ca6ea4","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-04-05 12:34:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1bfef55990e0182c83aee7f67292cb836a5543f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-04-12 14:59:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8db9cf11906a30e8df6b48984ce5f427e350fc7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-04-12 15:13:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5782ea9126d0bda05e4b385c0da32460a350459d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:22:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/151526977ff0419597eddb1446729023\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d3d043330549da9dad823abb5e179e : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde8a8e7bcdc4f1aa199341937c3bf40 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":3},{"id":"4117e965620ccdefb70bf5e3637166e538d61099","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-04-12 15:30:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c12708a993395fac7e376145dadfbcc62cc2c48c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-04-14 06:37:41.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"36e33581cb6e7724523a0cbc44044c5c45f3ef7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-04-14 10:02:05.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4002632ba35d0e90b5f54719f06184f7e35e12d7","author":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"date":"2023-04-14 10:03:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6208f57e08cc8181a05208f48a035665c3b72e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 10:13:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d0a01d267e54333ab7a09dc9f58c4f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5c6d00fcc047e29923c1e4bec49989 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9129b9ec5579415ca78518c8828ef053 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":4},{"id":"715c33405afb57e3983fdfbf980e09fc9b17df56","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-04-14 11:06:25.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a3b35bcabbb011ddf581ce33b72e06d5680457dd","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-04-26 13:55:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b599932d48620203b947f9f227fd3b05b4cbeb00","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-09 12:50:47.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be6ad21be45391a2567dbe6101a38c9e4c1c7167","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-21 05:09:12.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"17578b4cee720a6b61d536dae3324f5b100a21ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 05:09:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6567a8f756fad3f38970fa755d79b0f7622921da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 05:13:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0c39d0358984fe3bbe3b55dcbbe9252\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41a234920084395989e1d95e33bf832 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8734f946937a4217812518143966cda6 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":4},{"id":"4f543a8211c9b50c678b2f5554e61944ee0423f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 05:13:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b284316a90d170855179b6f91b90fbf999f15bba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 05:14:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f365d2155ccc4448b4ece1cee8dd94b6\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/918c8f0ac1554b5893068564823bee87 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"04076c9a63d171d05c29eae77aedae8dcb82ae5b","revisions":{"7ffc3989acd11148b5f65d72e4a06d310d4582c5":{"kind":"REWORK","_number":1,"created":"2023-03-13 08:39:02.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/30/877230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/877230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/1"}}},"commit":{"parents":[{"commit":"b3a1a0ddb65628df83cd27db0023efe46770a8e3","subject":"Spec folder for 2023.2 cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b3a1a0ddb65628df83cd27db0023efe46770a8e3"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-03-13 08:17:48.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-03-13 08:17:48.000000000","tz":60},"subject":"Re-propose extend volume completion action for 2023.2","message":"Re-propose extend volume completion action for 2023.2\n\nPreviously-approved: 2023.1\n\nFor previous discussion see the change at:\nhttps://review.opendev.org/c/openstack/cinder-specs/+/866718\n\nBlueprint: extend-volume-completion-action\nChange-Id: I40d1632953cac767c66f735a11e0477afd63426f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7ffc3989acd11148b5f65d72e4a06d310d4582c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/7ffc3989acd11148b5f65d72e4a06d310d4582c5"}]},"branch":"refs/heads/master"},"52eb688f1ce9030ef017a920f26898261d8b3143":{"kind":"REWORK","_number":2,"created":"2023-04-04 08:29:07.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/30/877230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/877230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/2"}}},"commit":{"parents":[{"commit":"b3a1a0ddb65628df83cd27db0023efe46770a8e3","subject":"Spec folder for 2023.2 cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b3a1a0ddb65628df83cd27db0023efe46770a8e3"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-03-13 08:17:48.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-04-04 08:28:50.000000000","tz":120},"subject":"Re-propose extend volume completion action for 2023.2","message":"Re-propose extend volume completion action for 2023.2\n\nPreviously-approved: 2023.1\n\nFor previous discussion see the change at:\nhttps://review.opendev.org/c/openstack/cinder-specs/+/866718\n\nBlueprint: extend-volume-completion-action\nChange-Id: I40d1632953cac767c66f735a11e0477afd63426f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52eb688f1ce9030ef017a920f26898261d8b3143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/52eb688f1ce9030ef017a920f26898261d8b3143"}]},"branch":"refs/heads/master"},"2165351792fbc91006cc6c870cd0125cb8743a79":{"kind":"REWORK","_number":3,"created":"2023-04-12 15:13:22.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/30/877230/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/877230/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/3"}}},"commit":{"parents":[{"commit":"b3a1a0ddb65628df83cd27db0023efe46770a8e3","subject":"Spec folder for 2023.2 cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b3a1a0ddb65628df83cd27db0023efe46770a8e3"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-03-13 08:17:48.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-04-12 15:12:33.000000000","tz":120},"subject":"Re-propose extend volume completion action for 2023.2","message":"Re-propose extend volume completion action for 2023.2\n\nPreviously-approved: 2023.1\n\nFor previous discussion see the change at:\nhttps://review.opendev.org/c/openstack/cinder-specs/+/866718\n\nBlueprint: extend-volume-completion-action\nChange-Id: I40d1632953cac767c66f735a11e0477afd63426f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2165351792fbc91006cc6c870cd0125cb8743a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/2165351792fbc91006cc6c870cd0125cb8743a79"}]},"branch":"refs/heads/master"},"04076c9a63d171d05c29eae77aedae8dcb82ae5b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-04-14 10:02:05.000000000","uploader":{"_account_id":34859,"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","username":"kgube"},"ref":"refs/changes/30/877230/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/30/877230/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/30/877230/4"}}},"commit":{"parents":[{"commit":"b3a1a0ddb65628df83cd27db0023efe46770a8e3","subject":"Spec folder for 2023.2 cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b3a1a0ddb65628df83cd27db0023efe46770a8e3"}]}],"author":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-03-13 08:17:48.000000000","tz":60},"committer":{"name":"Konrad Gube","email":"konrad.gube@cloudandheat.com","date":"2023-04-14 09:59:48.000000000","tz":120},"subject":"Re-propose extend volume completion action","message":"Re-propose extend volume completion action\n\nPreviously-approved: 2023.1\n\nFor previous discussion see the change at:\nhttps://review.opendev.org/c/openstack/cinder-specs/+/866718\n\nBlueprint: extend-volume-completion-action\nChange-Id: I40d1632953cac767c66f735a11e0477afd63426f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/04076c9a63d171d05c29eae77aedae8dcb82ae5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/04076c9a63d171d05c29eae77aedae8dcb82ae5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
