)]}'
{"id":"openstack%2Fcinder-specs~892576","triplet_id":"openstack%2Fcinder-specs~master~I72df8ed5b1d805e0596ba3d13ce28b29accedcae","project":"openstack/cinder-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-09-20 19:34:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72df8ed5b1d805e0596ba3d13ce28b29accedcae","subject":"Fix cinder CLI parameters for default-type-set","status":"MERGED","created":"2023-08-23 18:48:05.000000000","updated":"2023-09-20 19:36:24.000000000","submitted":"2023-09-20 19:34:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"892576","meta_rev_id":"c715c0010959bea1e3ea5846762750a48f047919","_number":892576,"virtual_id_number":892576,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-20 19:34:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":2,"date":"2023-09-20 19:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2023-08-29 18:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-08-30 05:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2023-09-20 19:30:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 18:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 18:36:29.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-08-30 05:35:35.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-09-20 19:30:12.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"}],"messages":[{"id":"ef10af8257af4666ac860022c3051611d71e8ceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-08-23 18:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57cc9dad8ee6ea3e6de2ed78ae49dae1d7b43b24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 18:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb8b79e0bae74db6b92eefdbdd811d7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f353322b87a548399a403b75d4bc7217 : FAILURE in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017c7cb1e257498d96c61cc43eb99fba : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c2fb4b0fe686b838c885c33d5c65001aa454ab3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-08-29 16:20:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"75b9f8f6d24c8905bc208b84df1083942195450d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 16:29:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11dff5cad34b45f68d7ac4a5a591a144\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3269f08fa224d308f702d50c5bcb7ee : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9627067ff07f4c48845e0a7932e8e5a8 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5e9aa40c659e528b8cf39d7cd46611b28f1769ac","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-08-29 18:36:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b3679e374f4a1fe5c8bea6b5e4843c16cf4b22f9","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-08-30 05:35:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"eaf3dd58d3cd638dcfcc684126f0fb4076e91c38","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-09-20 19:30:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ce2768cabd88dac622a0e740dea61162c0edaa5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 19:30:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"42855f1462f2179dde05bc9b41f78ef747d79659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 19:34:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4149a8fdbf164a6ba53a30c25ea75af9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07f7c8dfc6c48e4b00f13e231c9d528 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68a2f95a82b4dadad6a8756a3605471 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d2e241cd5e0156fcf3f3e278acd2ef5d8b95fe8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 19:34:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c715c0010959bea1e3ea5846762750a48f047919","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 19:36:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ea8dd9e6d4420cb73a660e28a58d14\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/742ec35548d64c97b501ec58f0c7205f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac0f9853b6d0b56b5c042909539304e2d79467c6","revisions":{"36dece5db380f0e96ccaf7cbc0dff4996140f25c":{"kind":"REWORK","_number":1,"created":"2023-08-23 18:48:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/76/892576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/76/892576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/1"}}},"commit":{"parents":[{"commit":"b3a1a0ddb65628df83cd27db0023efe46770a8e3","subject":"Spec folder for 2023.2 cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b3a1a0ddb65628df83cd27db0023efe46770a8e3"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-08-23 18:47:08.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-08-23 18:47:08.000000000","tz":-240},"subject":"Fix cinder CLI parameters for default-type-set","message":"Fix cinder CLI parameters for default-type-set\n\nThe arguments that the Cinder CLI takes are reversed\nfrom what this doc currently states.\n\n$ cinder --os-volume-api-version 3.62 default-type-set\nusage: cinder default-type-set \u003cvolume_type\u003e \u003cproject_id\u003e\nerror: the following arguments are required: \u003cvolume_type\u003e, \u003cproject_id\u003e\n\nChange-Id: I72df8ed5b1d805e0596ba3d13ce28b29accedcae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/36dece5db380f0e96ccaf7cbc0dff4996140f25c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/36dece5db380f0e96ccaf7cbc0dff4996140f25c"}]},"branch":"refs/heads/master"},"ac0f9853b6d0b56b5c042909539304e2d79467c6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-29 16:20:02.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/76/892576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/76/892576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/76/892576/2"}}},"commit":{"parents":[{"commit":"963ede4dd7d4095bc16ac6f1b171376ef874b70b","subject":"Require openstackdocstheme\u003e\u003d3.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/963ede4dd7d4095bc16ac6f1b171376ef874b70b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-08-23 18:47:08.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-08-29 16:20:02.000000000","tz":0},"subject":"Fix cinder CLI parameters for default-type-set","message":"Fix cinder CLI parameters for default-type-set\n\nThe arguments that the Cinder CLI takes are reversed\nfrom what this doc currently states.\n\n$ cinder --os-volume-api-version 3.62 default-type-set\nusage: cinder default-type-set \u003cvolume_type\u003e \u003cproject_id\u003e\nerror: the following arguments are required: \u003cvolume_type\u003e, \u003cproject_id\u003e\n\nChange-Id: I72df8ed5b1d805e0596ba3d13ce28b29accedcae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ac0f9853b6d0b56b5c042909539304e2d79467c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/ac0f9853b6d0b56b5c042909539304e2d79467c6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
