)]}'
{"id":"openstack%2Fcinder-specs~921366","triplet_id":"openstack%2Fcinder-specs~master~I7536015287109479c473e22cc4c3da42921ec7c7","project":"openstack/cinder-specs","branch":"master","attention_set":{"30748":{"account":{"_account_id":30748,"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","username":"marek.chmiel"},"last_update":"2024-06-05 10:03:28.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7536015287109479c473e22cc4c3da42921ec7c7","subject":"Spec for backup backend switch","status":"NEW","created":"2024-06-05 09:53:46.000000000","updated":"2026-06-25 14:03:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"000aec5403b4cfb5d6cb413831c14f11f8d2747f","_number":921366,"virtual_id_number":921366,"owner":{"_account_id":30748,"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","username":"marek.chmiel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-05 10:32:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"disliked":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"all":[{"value":-1,"date":"2025-12-05 14:32:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":-2,"date":"2026-06-25 14:03:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"}],"CC":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-05 10:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 13:30:14.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2024-11-15 14:54:23.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"CC"},{"updated":"2025-12-05 14:32:01.000000000","updated_by":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"REVIEWER"}],"messages":[{"id":"ee4ff77086f204e1e544b8ca24ae09998531aa9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30748,"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","username":"marek.chmiel"},"date":"2024-06-05 09:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0fc3c8801f9e9282e4e32f61d3c520312709b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 10:03:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9011c026fa8c4310a55d05e9c65a5300\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6796e56373674824b06ee5c9b0aef47f : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a493b204f395459994484e30dd476da7 : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1179b59fbf1ef17a4ec5ddfd3a8129b61615fd59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30748,"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","username":"marek.chmiel"},"date":"2024-06-05 10:26:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"436d6c8bf7522856b27b53a5274479efc416a3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 10:32:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01be2fe4628549dcac08cdd9ed4f6abf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8565e7b0b5424af788277e548b1b79eb : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1847fc61ed414493b3e81c08290aaba4 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b8ce508fc320f53c3acccd52721f455b937dd190","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-01 13:30:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b66a563fa29b37be5c83b8e04bdd39842f45476","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-12-05 14:32:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"000aec5403b4cfb5d6cb413831c14f11f8d2747f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-06-25 14:03:24.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"23864da0f3f8bec98f8e35d2acb5e0a781e460e7","revisions":{"d7bff7d81f1bcbb092608d75d6199baa8bb3a572":{"kind":"REWORK","_number":1,"created":"2024-06-05 09:53:46.000000000","uploader":{"_account_id":30748,"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","username":"marek.chmiel"},"ref":"refs/changes/66/921366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/921366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/1"}}},"commit":{"parents":[{"commit":"3fe68c9fad63d7c577787d500155a6016d771e0f","subject":"Add spec folder for 2024.2 (Dalmation)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3fe68c9fad63d7c577787d500155a6016d771e0f"}]}],"author":{"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","date":"2024-06-05 09:50:46.000000000","tz":120},"committer":{"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","date":"2024-06-05 09:50:46.000000000","tz":120},"subject":"Spec for backup backend switch","message":"Spec for backup backend switch\n\nThere should be a way to switch to a different backup backend on existing\ncinder deployment, where data is already stored without losing the data.\n\nThe idea is to transition from old to new backup backend as transparently as\npossible from end-user perspective, without the need to move data between\nbackends, or any action by end-user.\n\nOnce the switch is done, no new data should be written to old backend.\n\nPartial-Implements: https://blueprints.launchpad.net/cinder/+spec/backup-backend-switch\nChange-Id: I7536015287109479c473e22cc4c3da42921ec7c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d7bff7d81f1bcbb092608d75d6199baa8bb3a572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/d7bff7d81f1bcbb092608d75d6199baa8bb3a572"}]},"branch":"refs/heads/master"},"23864da0f3f8bec98f8e35d2acb5e0a781e460e7":{"kind":"REWORK","_number":2,"created":"2024-06-05 10:26:28.000000000","uploader":{"_account_id":30748,"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","username":"marek.chmiel"},"ref":"refs/changes/66/921366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/66/921366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/66/921366/2"}}},"commit":{"parents":[{"commit":"3fe68c9fad63d7c577787d500155a6016d771e0f","subject":"Add spec folder for 2024.2 (Dalmation)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/3fe68c9fad63d7c577787d500155a6016d771e0f"}]}],"author":{"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","date":"2024-06-05 09:50:46.000000000","tz":120},"committer":{"name":"Marek Chmiel","email":"marek.chmiel@ovhcloud.com","date":"2024-06-05 10:26:20.000000000","tz":120},"subject":"Spec for backup backend switch","message":"Spec for backup backend switch\n\nThere should be a way to switch to a different backup backend on existing\ncinder deployment, where data is already stored without losing the data.\n\nThe idea is to transition from old to new backup backend as transparently as\npossible from end-user perspective, without the need to move data between\nbackends, or any action by end-user.\n\nOnce the switch is done, no new data should be written to old backend.\n\nPartial-Implements: https://blueprints.launchpad.net/cinder/+spec/backup-backend-switch\nChange-Id: I7536015287109479c473e22cc4c3da42921ec7c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/23864da0f3f8bec98f8e35d2acb5e0a781e460e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/23864da0f3f8bec98f8e35d2acb5e0a781e460e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
