)]}'
{"id":"openstack%2Fcinder-specs~989322","triplet_id":"openstack%2Fcinder-specs~master~I7b30409119115eeb2e9822e35a2a562b456214b2","project":"openstack/cinder-specs","branch":"master","topic":"fqi-dev","attention_set":{"34260":{"account":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"last_update":"2026-09-23 15:25:36.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-09-18 02:11:54.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"38251":{"account":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"last_update":"2026-08-21 20:44:22.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-07-16 03:05:02.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-06-24 10:20:48.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-09-23 15:25:36.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-06-05 04:46:14.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2026-05-27 03:29:50.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"I7b30409119115eeb2e9822e35a2a562b456214b2","subject":"Add spec for retype volume including snapshots","status":"NEW","created":"2026-05-20 03:33:02.000000000","updated":"2026-10-05 19:56:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":96,"unresolved_comment_count":38,"has_review_started":true,"meta_rev_id":"82cdb7afc7c59e7da9c4c528c2b475d318a01387","_number":989322,"virtual_id_number":989322,"owner":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-30 04:16:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2026-10-05 19:56:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 03:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-20 16:01:15.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2026-05-20 20:31:03.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"CC"},{"updated":"2026-05-27 03:29:19.000000000","updated_by":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2026-05-27 03:29:50.000000000","updated_by":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"CC"},{"updated":"2026-06-04 20:36:32.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"CC"},{"updated":"2026-06-05 14:50:37.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-06-05 15:19:34.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-24 15:23:52.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-06-30 06:22:04.000000000","updated_by":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"reviewer":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"state":"CC"}],"messages":[{"id":"d4bfa08ee11fe06cb5ed36bd43640b57eeb0c51b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-05-20 03:33:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5507c74ded4532675382233a38533ad8dee9408a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 03:41:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9654049814ba4da7953262e2c8b8decb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec587f919ff4020b4054aa3c07493b1 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c8cd8a87734ee599d209884d6c4910 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2516ac3ecaf517039a1867da0392c720f8ea7956","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-05-20 16:01:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4b3f33806e3fa3d3dd883f1e4fa07625e7f74188","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2026-05-20 20:31:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2434bbe4e111d00e32a51c98071dd40d839530a2","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-05-27 03:27:26.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e28e63b7e6c66e6b2af356861cd5ce0520050eae","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-05-27 03:29:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79ad9cb312c117c38a67c39e641e94e7965601ad","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-05-28 01:48:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1978c91b2f267f13144e48574e0ea6c65a129ed","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-06-04 20:36:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e605df5f4738f324e479a09452725202a7ffb5f5","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-06-05 04:27:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"821179211564215472ae59cf3dfff1a36c78610c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-06-05 04:35:38.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"addb19ab5b54b87bbe8cd4cc4ae9912e26a210cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 04:39:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1581fef9e2b4dd490d2b7d6c8d9eead\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e3f1de8a6c44a7ab6228f2d8de6a7a : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34979fe2269143519667c1199ed2b1b2 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2306b245c62dc5cdf77ff0ebee1b6ea0cd110e98","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-06-05 04:46:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fdfa12de730855815ad749176ec095aa20bcb99e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-06-05 08:53:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"383d6fec39d28bc9c652b7f054e5c190c36e9e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 08:59:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2d82b2d4c1c4c37b903de9b1667bad6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252d8c2f54d04d3ab31d7df5718c7017 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10ecf28f25f413d957a645947ec1cd9 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"0bd141b85751a7a8e6e16bac99a5b10f71eab358","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-06-05 13:28:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0eacc5ac1d0235a7d1d03370e583e2c37b494e30","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-06-05 14:50:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"abd5e1bfbd40d351dd440105e115e6495d1c1767","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-08 14:50:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4736281341cf71bee828c73a3c3b19257eddaf89","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-06-09 12:36:16.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e8db03869e3e3c03bc1d06dcb7886013e5debf06","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-11 12:13:06.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9eb4a2e1567a59174f95232343f56b6839422bed","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-06-15 01:21:19.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3c7a5092a5815268ecda936b360e6a8aec26c05c","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-15 05:17:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b1c3c89c7f7ebf80cbc6b8f74cfa5e97403da4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-06-24 10:20:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6dee563798866830a0a22a4c637fe2d88b201b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:46:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/233dfb40454d4a1baba3548db43ae297\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060c2753b0264de98b59e090a70f3fe6 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bb4a23add44c23bceaae6538f90873 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":4},{"id":"776c5333bebab69e327649093c5af963394489ef","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-24 15:23:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9b3f77b27f25f3c721884f0cec66e3f6b042dd08","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-25 14:48:42.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1f4905b1485317969196174bafecbd65f0d85b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-06-30 06:17:55.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"dd628c9ef99bbfe7ca0f1af4858a36d8468a0857","author":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"date":"2026-06-30 06:22:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e42f1360d2d1a4e3e150e685da5aba7fd2f17e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 06:33:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a65b7a9d4bf4f40ab11d4be170c3f75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e37dec0d68430e9539bd06905681f8 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b53450a3bd94d5f95e3fec1b825bb6c : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":5},{"id":"cc185d76e361d6f144105d126d4b2e867e072f5a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-30 14:15:17.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"486aea99deea7e8b91a87512900b956e24030d4e","author":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"date":"2026-07-01 00:38:21.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b898e8b7c11b522244d8f7ecea1f70ea094b44cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-07-01 02:54:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5e6cc288c5aefdf6965e20641914e464942477f2","author":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"date":"2026-07-01 02:56:48.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9c159020890aced9ed8b92b06889266de37c3886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 03:14:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/786bfdf7b1314acb9e63a0b1e54507a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b996e6780c34032a2e4fa95e5bb1eda : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22edf1e907424ed4bba7202c3a6f74cd : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":6},{"id":"097b5e346f9b4ca30e7621c90ab96d3b0b0d2011","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-09 13:43:50.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7cbad3830e52e4d8fa3563bdaa7d4db6bf725b0a","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-07-13 22:39:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9aafe287061d386d55317992ca51567190871068","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-07-14 22:29:02.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8bab9fb650af59b95f4dd7f7f8d098449c3a3da2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-07-16 03:05:02.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"d968ac4b8a47de5add4a7dd5f6433cfa47955a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 03:32:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d2981b2e2240ceb76e580f5b95d89d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ae3ae5fb3247819a8dfca118d8d9ca : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bca254cd6db40af85f879cc8767da60 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":7},{"id":"315f58cb89e892c3a1fa19d8b44db814292a602b","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-12 13:55:55.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ee82b777bf93a4c410a13d8c3b1293250395fb38","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-12 21:41:34.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ed12b197f09a16f8200cc49a58feac617111d3eb","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-08-18 19:13:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"83acfffb62b4c210410e5f4c3f3c32042135f220","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-21 20:44:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8638c1371c0144ae1e8d6a474b16a12aac05f672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-09-18 02:11:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"233c87bc9fb3c92fb520e15ea474da2e202f7f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 03:00:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ba0f12acd1430fb41eafec74ba7bc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbcc010cfb949468e8dbfbd0621aa2e : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2111ada3ce524a0baff268c81b98eae2 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":8},{"id":"78c9ecdfe79394f929ce5538d02177f8d9a2d40f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-23 15:25:36.000000000","message":"Patch Set 8:\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0979e21bf952b80ee534ec11e373f5e9cda700f4","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-23 18:00:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"51a7035eeb5b17953f7c3ad585fc9a41a580dfc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"date":"2026-09-30 04:10:07.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"2f14b4f809d0e6861db247dc00ef2f110c71ed77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 04:16:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ecac24a48fb4a978f53a5867ed7b9d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a48fddc00e441f90be10f82f585981 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5011291a5a485a88748756e5e1f401 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":9},{"id":"23d6e4ee9398b22dbeafbd74dc9d5eda13d65eca","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-10-05 19:55:56.000000000","message":"Patch Set 9:\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"82cdb7afc7c59e7da9c4c528c2b475d318a01387","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-10-05 19:56:04.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"40f4a1267adaf71cd6a2426acbfabf74204d1e05","revisions":{"1c8a542f0d9e8a3b6b8184de497b691f834c6abe":{"kind":"REWORK","_number":1,"created":"2026-05-20 03:33:02.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/1"}}},"commit":{"parents":[{"commit":"63ef47423a1c6703b872dc3d44fad6fd2dcf8a25","subject":"Merge \"Remove url tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/63ef47423a1c6703b872dc3d44fad6fd2dcf8a25"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-05-20 03:19:50.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-05-20 03:29:22.000000000","tz":540},"subject":"This specification proposes an enhancement to Cinder volume migration to allow migrating an available or in-use volume together with all of its snapshots between different backends, preserving the lineage of snapshots.","message":"This specification proposes an enhancement to Cinder volume migration\nto allow migrating an available or in-use volume together with all of\nits snapshots between different backends, preserving the lineage of\nsnapshots.\n\nBlueprint: volume-migration-with-snapshot\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1c8a542f0d9e8a3b6b8184de497b691f834c6abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/1c8a542f0d9e8a3b6b8184de497b691f834c6abe"}]},"branch":"refs/heads/master"},"276ec0da9226539c599ed4a32e89af575a61adce":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-05 04:35:38.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/2"}}},"commit":{"parents":[{"commit":"63ef47423a1c6703b872dc3d44fad6fd2dcf8a25","subject":"Merge \"Remove url tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/63ef47423a1c6703b872dc3d44fad6fd2dcf8a25"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-05-20 03:19:50.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-05 03:51:22.000000000","tz":540},"subject":"This specification proposes an enhancement to Cinder volume migration to allow migrating an available or in-use volume together with all of its snapshots between different backends, preserving the lineage of snapshots.","message":"This specification proposes an enhancement to Cinder volume migration to allow migrating an available or in-use volume together with all of its snapshots between different backends, preserving the lineage of snapshots.\n\nBlueprint: volume-migration-with-snapshot\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/276ec0da9226539c599ed4a32e89af575a61adce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/276ec0da9226539c599ed4a32e89af575a61adce"}]},"branch":"refs/heads/master"},"6a25305a858607d447ff1747757ad84fce51eec2":{"kind":"REWORK","_number":3,"created":"2026-06-05 08:53:36.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/3"}}},"commit":{"parents":[{"commit":"63ef47423a1c6703b872dc3d44fad6fd2dcf8a25","subject":"Merge \"Remove url tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/63ef47423a1c6703b872dc3d44fad6fd2dcf8a25"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-05 08:53:03.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-05 08:53:22.000000000","tz":540},"subject":"Add spec for retype volume including snapshots","message":"Add spec for retype volume including snapshots\n\nThis specification proposes an enhancement to Cinder retype volume to\nallow migrating a volume that includes snapshots between different\nbackends with different volume types, preserving the lineage of\nsnapshots.\n\nBlueprint: retype-volume-including-snapshots\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6a25305a858607d447ff1747757ad84fce51eec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6a25305a858607d447ff1747757ad84fce51eec2"}]},"branch":"refs/heads/master"},"66c49be9b45587a93f816a5a40e8930257d27055":{"kind":"REWORK","_number":4,"created":"2026-06-24 10:20:48.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/4"}}},"commit":{"parents":[{"commit":"63ef47423a1c6703b872dc3d44fad6fd2dcf8a25","subject":"Merge \"Remove url tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/63ef47423a1c6703b872dc3d44fad6fd2dcf8a25"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-24 10:18:56.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-24 10:19:40.000000000","tz":540},"subject":"Add spec for retype volume including snapshots","message":"Add spec for retype volume including snapshots\n\nThis specification proposes an enhancement to Cinder retype volume to\nallow migrating a volume that includes snapshots between different\nbackends with different volume types, preserving the lineage of\nsnapshots.\n\nBlueprint: retype-volume-including-snapshots\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/66c49be9b45587a93f816a5a40e8930257d27055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/66c49be9b45587a93f816a5a40e8930257d27055"}]},"branch":"refs/heads/master"},"6bd9dec9e82f7f3a92a78dba41e1cb74e58ab662":{"kind":"REWORK","_number":5,"created":"2026-06-30 06:17:55.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/5"}}},"commit":{"parents":[{"commit":"63ef47423a1c6703b872dc3d44fad6fd2dcf8a25","subject":"Merge \"Remove url tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/63ef47423a1c6703b872dc3d44fad6fd2dcf8a25"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-24 10:18:56.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-30 06:19:46.000000000","tz":540},"subject":"Add spec for retype volume including snapshots","message":"Add spec for retype volume including snapshots\n\nThis specification proposes an enhancement to Cinder retype volume to\nallow migrating a volume that includes snapshots between different\nbackends with different volume types, preserving the lineage of\nsnapshots.\n\nBlueprint: retype-volume-including-snapshots\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6bd9dec9e82f7f3a92a78dba41e1cb74e58ab662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/6bd9dec9e82f7f3a92a78dba41e1cb74e58ab662"}]},"branch":"refs/heads/master"},"23291dce6a95a84d65ce0a72711488eee1d3bed4":{"kind":"REWORK","_number":6,"created":"2026-07-01 02:54:55.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/6"}}},"commit":{"parents":[{"commit":"63ef47423a1c6703b872dc3d44fad6fd2dcf8a25","subject":"Merge \"Remove url tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/63ef47423a1c6703b872dc3d44fad6fd2dcf8a25"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-06-24 10:18:56.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-07-01 02:56:53.000000000","tz":540},"subject":"Add spec for retype volume including snapshots","message":"Add spec for retype volume including snapshots\n\nThis specification proposes an enhancement to Cinder retype volume to\nallow migrating a volume that includes snapshots between different\nbackends with different volume types, preserving the lineage of\nsnapshots.\n\nBlueprint: retype-volume-including-snapshots\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/23291dce6a95a84d65ce0a72711488eee1d3bed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/23291dce6a95a84d65ce0a72711488eee1d3bed4"}]},"branch":"refs/heads/master"},"c2e62e6502565c944c7dd9ea47e0946b303ebbc1":{"kind":"REWORK","_number":7,"created":"2026-07-16 03:05:02.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/7"}}},"commit":{"parents":[{"commit":"b3f580471e792c8780168ac7736a6326c7ab241e","subject":"Add spec for backup container name templates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b3f580471e792c8780168ac7736a6326c7ab241e"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-07-16 02:55:40.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-07-16 02:55:40.000000000","tz":540},"subject":"Add spec for retype volume including snapshots","message":"Add spec for retype volume including snapshots\n\nThis specification proposes an enhancement to Cinder retype volume to\nallow migrating a volume that includes snapshots between different\nbackends with different volume types, preserving the lineage of\nsnapshots.\n\nBlueprint: retype-volume-including-snapshots\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c2e62e6502565c944c7dd9ea47e0946b303ebbc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/c2e62e6502565c944c7dd9ea47e0946b303ebbc1"}]},"branch":"refs/heads/master"},"41371d77ccc0932439468b3ebb02491621df45d1":{"kind":"REWORK","_number":8,"created":"2026-09-18 02:11:54.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/8"}}},"commit":{"parents":[{"commit":"b3f580471e792c8780168ac7736a6326c7ab241e","subject":"Add spec for backup container name templates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/b3f580471e792c8780168ac7736a6326c7ab241e"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-09-18 02:08:39.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-09-18 02:08:52.000000000","tz":540},"subject":"Add spec for retype volume including snapshots","message":"Add spec for retype volume including snapshots\n\nThis specification proposes an enhancement to Cinder retype volume to\nallow migrating a volume that includes snapshots between different\nbackends with different volume types, preserving the lineage of\nsnapshots.\n\nBlueprint: retype-volume-including-snapshots\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/41371d77ccc0932439468b3ebb02491621df45d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/41371d77ccc0932439468b3ebb02491621df45d1"}]},"branch":"refs/heads/master"},"40f4a1267adaf71cd6a2426acbfabf74204d1e05":{"kind":"REWORK","_number":9,"created":"2026-09-30 04:10:07.000000000","uploader":{"_account_id":38251,"name":"Fuli Qi","display_name":"Fuli Qi","email":"qi.fuli@fujitsu.com","username":"fqi"},"ref":"refs/changes/22/989322/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder-specs","ref":"refs/changes/22/989322/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder-specs refs/changes/22/989322/9"}}},"commit":{"parents":[{"commit":"31cde8f8ad66da78fbe9eebc4123ece5a7a6552d","subject":"Add 2027.1 spec structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/31cde8f8ad66da78fbe9eebc4123ece5a7a6552d"}]}],"author":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-09-30 01:01:50.000000000","tz":540},"committer":{"name":"Fuli Qi","email":"qi.fuli@fujitsu.com","date":"2026-09-30 01:04:22.000000000","tz":540},"subject":"Add spec for retype volume including snapshots","message":"Add spec for retype volume including snapshots\n\nThis specification proposes an enhancement to Cinder retype volume to\nallow migrating a volume that includes snapshots between different\nbackends with different volume types, preserving the lineage of\nsnapshots.\n\nBlueprint: retype-volume-including-snapshots\n\nChange-Id: I7b30409119115eeb2e9822e35a2a562b456214b2\nSigned-off-by: Fuli Qi \u003cqi.fuli@fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/40f4a1267adaf71cd6a2426acbfabf74204d1e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder-specs/commit/40f4a1267adaf71cd6a2426acbfabf74204d1e05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
